)]}'
{"id":"openstack%2Fplacement~963808","triplet_id":"openstack%2Fplacement~master~I569e7f553edc890a701ac6599a8218076f13d2bd","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-18 19:44:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I569e7f553edc890a701ac6599a8218076f13d2bd","subject":"Replace remaining py39 target","status":"MERGED","created":"2025-10-12 17:59:32.000000000","updated":"2025-11-18 19:45:17.000000000","submitted":"2025-11-18 19:44:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963808","meta_rev_id":"a8412f6e8faa8888981a1c11c285938e669394f6","_number":963808,"virtual_id_number":963808,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 19:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-18 18:49:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-18 18:49:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-12 19:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 18:49:03.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"0ebc64c1e37301ad8f16a345781768faa32ff9ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-12 17:59:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dae97369af5d3b09edcdbc9c17f87ddd1c0c019a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-12 19:12:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18b801cfa5a046b499b951f6a0bcc82d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f8bed91338314b5aa1a4ffa65281e2f5 : SUCCESS in 4m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a97256078c6542ee9f9b207ec8427de2 : SUCCESS in 1h 03m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c7d582db871434eb7fb9ca3b6a98965 : SUCCESS in 1h 07m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f369312097420fb3bf5459e2c2da50 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8299c30e2ee46399bf1936088359e8b : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd3c639e624049c5a193ac2144c9b7ae : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e536d1f0dcf4a20b8fb1d3b60ca5a71 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f919f11d525499ca3d3554802b8375a : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212e24cdb91445188d3adb156c15cb65 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49d73359faff46d3950c92d89fcc4e99 : SUCCESS in 5m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f1ee7b56714428f9014828cbb3dc57f : SUCCESS in 7m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/62159016c32a42c88055bf690139f6fa : SUCCESS in 7m 08s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9a924dd686664f2c9a12528e89e84371 : SUCCESS in 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4412dc20f30561bd3a16de896d1f50d10a4dd268","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-18 18:49:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45813a388c6c48975ee393243b31044a7dc2d2f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 18:49:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bcfab15fc2b9619d2ba70cde20e7343f6335c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 19:44:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf58c7d7dae948e4b2a26a68ce5e56b3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b227936efe2346db8efc3f7320435221 : SUCCESS in 4m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7bcca353fbd9460ea813bd00ec1f8784 : SUCCESS in 32m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd15c6b50c014b7a92cac53c0c83d320 : SUCCESS in 50m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003478a524b841c4a809a9e51662c8e7 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ed36dcc7ad143a185225c61b5b1779b : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cbcf27f30634d94b5260dc858e41cee : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7ebc54ddae44a59644b448870a2f14 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dcd59a11cb74fb5bfb49e03f2a992d5 : SUCCESS in 4m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f84982e82cb342e29b2a5bad893ce360 : SUCCESS in 6m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ad40d884d5f94012bc57451907d35e96 : SUCCESS in 7m 19s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eff94e07172e4bc9b001928de966fe83 : SUCCESS in 23m 51s","accounts_in_message":[],"_revision_number":1},{"id":"35516e5166b95be606dca697b7f5f9352c0dd0b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 19:44:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8412f6e8faa8888981a1c11c285938e669394f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 19:45:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab284690a5964dd59718e7fff30af18e\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0fdfbb24636e4fa393b52224abb97ee9 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507062947d1949819758f9cddfee10aa : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dae309697ae4ee2a24c89988c53571d : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cda3dbc3a07c2dbe3e31e7eef2bece369466d424","revisions":{"cda3dbc3a07c2dbe3e31e7eef2bece369466d424":{"kind":"REWORK","_number":1,"created":"2025-10-12 17:59:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/963808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/08/963808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/08/963808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/08/963808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/08/963808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/08/963808/1"}}},"commit":{"parents":[{"commit":"89bf1a232f102a658e568652f9cfd6f102aaabef","subject":"Remove excessive logging from GET a_c","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/89bf1a232f102a658e568652f9cfd6f102aaabef"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-12 17:59:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-12 17:59:22.000000000","tz":540},"subject":"Replace remaining py39 target","message":"Replace remaining py39 target\n\n... because Python 3.9 is no longer supported.\n\nChange-Id: I569e7f553edc890a701ac6599a8218076f13d2bd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cda3dbc3a07c2dbe3e31e7eef2bece369466d424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cda3dbc3a07c2dbe3e31e7eef2bece369466d424"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
