)]}'
{"id":"openstack%2Fplacement~967600","triplet_id":"openstack%2Fplacement~master~Id419bded99010378ed509163593efdfc91126037","project":"openstack/placement","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-25 14:02:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id419bded99010378ed509163593efdfc91126037","subject":"Correct default oslo.conf file","status":"MERGED","created":"2025-11-18 19:00:44.000000000","updated":"2025-11-25 17:23:03.000000000","submitted":"2025-11-25 14:02:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967600-pyproject","meta_rev_id":"723624cf34721f25038d4346bf5d16053a88519e","_number":967600,"virtual_id_number":967600,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 14:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-11-25 11:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-11-25 12:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-11-25 12:06:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 20:51:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-25 11:47:04.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-11-25 12:06:40.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"4075428a2814c49bd6918b1a33d4685dba50e969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-18 19:00:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c40c18e0e78134402593b17dc3851b7897dcc2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:51:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bfe5f1fb03142c5b126588bdaf00caa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/79257689f381453f8eedfae5bbb7d0e3 : SUCCESS in 2m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e31a00b1cb2a414da1da140543a4dd68 : SUCCESS in 43m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a669c62de2e41cc966cbff1a77b639e : SUCCESS in 29m 58s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/2aadc9e082604e75b85d8183f2cf9daf : SUCCESS in 1h 43m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67173f0e17a346d89a94578ba667df74 : SUCCESS in 1h 02m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b108151e237456da3cacb3dec13dd6f : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb13576718143ffa2e21e7cea05c014 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f579751ba43f4ecaa1f213a9b47f2559 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e7303973de04d6aaf1fd63a596b01cd : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a33f9054dc49c0baf96f4250f75f80 : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13bc204417c14190ac677e05c29570d4 : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0e6dde27abd54e61a6a571341aa640ab : SUCCESS in 10m 11s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0e622d5ec1eb4e4da3adc5d5c8f1da90 : SUCCESS in 25m 51s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/f9d0335f44c2406593942d53202d55be : FAILURE in 9m 40s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/2324d8045d13458b9c726505f215dbb7 : FAILURE in 3m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab4caa29293d496391a45da2d58709ab : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"925b0d8d3390f033a7b46731cd720d6e01d0865d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-25 11:47:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4907ceba5f79f6c6f134fc62cc534d5bd585071c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-25 12:06:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54a2ba7ec597054027ae7f48939c769acff6c62c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 12:07:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cd85b5dcfc2b2365157fda6fbf06229683cd3ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e9e772636f4667bde297f0918b9b00\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b7b95de96a0d43c19fddbe749e195b8e : SUCCESS in 2m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a3930a64fe7b4596ab491703d763ae14 : SUCCESS in 35m 00s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/db95dfd15b0c4134bd079937519c4972 : SUCCESS in 1h 45m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d85a3dcccb064cd481fea138d53e74d5 : SUCCESS in 33m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/589ba4448b70461f828e997d181582be : SUCCESS in 50m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73cbaac55fe142b99c4ba9c0d909afce : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d55d4356ac124972b13e4a036afef1eb : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75806bc4de91455d90e979135773c374 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72abc711e8e475a8e9e928374be1e9a : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d600dbda69240c5890cee3d5b89149b : SUCCESS in 4m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/039e642f026949c68e67614220b5e0ff : SUCCESS in 12m 01s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e51e039876524ccf9f3c8e408ac0c096 : SUCCESS in 20m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb7a6418b8c94d1496f90e3c30a40fdc : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"769e320756de898e52ec3ac8c11dd74b7a6de37e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"723624cf34721f25038d4346bf5d16053a88519e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-25 17:23:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5990908b7a292d86342b0d3d414f44e3e55aae1c","revisions":{"5990908b7a292d86342b0d3d414f44e3e55aae1c":{"kind":"REWORK","_number":1,"created":"2025-11-18 19:00:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/00/967600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/00/967600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/00/967600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/00/967600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/00/967600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/00/967600/1"}}},"commit":{"parents":[{"commit":"39f4895d9c5361ea529180d68f8bf92fcba1fd2a","subject":"setup: Remove pbr\u0027s wsgi_scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/39f4895d9c5361ea529180d68f8bf92fcba1fd2a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:57:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:57:23.000000000","tz":0},"subject":"Correct default oslo.conf file","message":"Correct default oslo.conf file\n\nClearly no one is using this. Correct it all the same.\n\nChange-Id: Id419bded99010378ed509163593efdfc91126037\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5990908b7a292d86342b0d3d414f44e3e55aae1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5990908b7a292d86342b0d3d414f44e3e55aae1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
