)]}'
{"id":"openstack%2Fplacement~967601","triplet_id":"openstack%2Fplacement~master~Ied4c4e6382b61119e15635606834f9f399ec24cf","project":"openstack/placement","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-25 14:02:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied4c4e6382b61119e15635606834f9f399ec24cf","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2025-11-18 19:00:44.000000000","updated":"2025-11-25 17:22:58.000000000","submitted":"2025-11-25 14:02:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967601-pyproject","meta_rev_id":"b19b0312d169b1f819cb8acfea495b5e1069873c","_number":967601,"virtual_id_number":967601,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 14:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-11-25 11:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-11-25 12:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-11-25 12:07:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 20:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-25 11:49:25.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-11-25 12:07:32.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"5445f5d25ddb76fbb95c80633a93f37bfca89ee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-18 19:00:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8d74f397ee92faec6b7f19015ac17dbf748a089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:49:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bacc785f0344f6eab42f128a74ec645\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2fd15a21f33345b29d631d03641347d6 : SUCCESS in 1m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a73a18786fb742b28edd6b662ec75ecf : SUCCESS in 2m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/dc0abc4e83b64f619b2a59d97c3d3d37 : SUCCESS in 30m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5bc3730446b9427d91c988b80beb42d3 : SUCCESS in 57m 41s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/4a8c738252f94783b07d0029f68b3f36 : SUCCESS in 1h 43m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a8cce9d00354818b9167f273f7a2e72 : SUCCESS in 1h 00m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d8af9a31cde46d9ab37e6c77b768001 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f79043dc83941eba93937f9575e4eca : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bec82d1c4bc44cc98b9cae1e50209f4 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84488a585a9c4c5e876347c66eccdf0a : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6c456e0e4043159ef00995203b7a0d : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c4060c9515b43349201c280e7e0d445 : SUCCESS in 3m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/340fdd5fe60b43349179fbc35fe62535 : SUCCESS in 8m 03s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f93df0b7acbc4183bf3490144eca2f65 : SUCCESS in 22m 40s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e6777d5f26a14cd0be0dccde600d1305 : FAILURE in 8m 30s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/612f02080cc342818329eb6400e6658a : FAILURE in 3m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bed4dcb6e66d4e44a4e138ee66f15cb9 : SUCCESS in 1h 06m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3653035d5798d49b1a4fd00c0b017d5f7a04dc02","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-25 11:49:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6709a515a14cd029c1b79e6afa13023a3015862f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-25 12:07:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"588d27e9e998aac9bae81752a4f8877dd0a08f61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 12:08:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b37abcb4b98255dcc0b36d4e40b02f8f2f3fbe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f42366cf60a945a9974d145d4bdd5e8f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e4a71a4853d24c59b3bd0eaf922cf49c : SUCCESS in 3m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/89e85e8ce0de47ce96cd672d0cc52698 : SUCCESS in 4m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/6b4b262704764558a600239a01c9b988 : SUCCESS in 1h 03m 04s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/6fb94643ff624928beeff8044f3a429a : SUCCESS in 59m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db31cbfb77a049269c5474c84f5af7a6 : SUCCESS in 59m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cffb06879d64365b6c7f3a9ad9b87cc : SUCCESS in 1h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b92fe8ac8784523a8b4316786454888 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5468e0da1a0475ea65109ff4d100241 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dde0d034ec641feb2a5ce6846d71c6f : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b2f2d6c6cb485d8374f0111b5bbe37 : SUCCESS in 5m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38414ef431b44284bb9cc25f8ee627e3 : SUCCESS in 3m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/535591111b1f4fce98ba5b51625d8de9 : SUCCESS in 8m 41s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1346e6e031224999ac4c522816b6e01d : SUCCESS in 23m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a539e0d612249639170f1a9265d8be9 : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8862a21cc07cd110e689061dd0930a8be7753525","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84491f99f4f59ae0ab59a099c6255afe841fdc04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:04:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05dd386c5f4b4710bd87a17b1b7dc5fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61583835525b4d328faba53bf4ba6f86 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"b19b0312d169b1f819cb8acfea495b5e1069873c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-25 17:22:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99039e66c898a1a33c0f1b3cc9a220925adff984","revisions":{"99039e66c898a1a33c0f1b3cc9a220925adff984":{"kind":"REWORK","_number":1,"created":"2025-11-18 19:00:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/967601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/01/967601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/01/967601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/01/967601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/01/967601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/01/967601/1"}}},"commit":{"parents":[{"commit":"5990908b7a292d86342b0d3d414f44e3e55aae1c","subject":"Correct default oslo.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5990908b7a292d86342b0d3d414f44e3e55aae1c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:56:30.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:57:37.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Ied4c4e6382b61119e15635606834f9f399ec24cf\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/99039e66c898a1a33c0f1b3cc9a220925adff984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/99039e66c898a1a33c0f1b3cc9a220925adff984"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
