)]}'
{"id":"openstack%2Fplacement~967602","triplet_id":"openstack%2Fplacement~master~Ib0d97b69f513439e62a6e734ba4e278510a5b670","project":"openstack/placement","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-11-25 16:14:44.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-25 16:14:44.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-11-25 15:08:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"Ib0d97b69f513439e62a6e734ba4e278510a5b670","subject":"pre-commit: Bump versions","status":"MERGED","created":"2025-11-18 19:00:44.000000000","updated":"2025-11-25 16:15:41.000000000","submitted":"2025-11-25 16:14:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967602-pyproject","meta_rev_id":"f9d36a2955bda720a953d7030988991e51a90ed4","_number":967602,"virtual_id_number":967602,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 16:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-25 15:08:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-25 15:08:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 20:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-25 11:50:49.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-11-25 12:14:15.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"5b98e5469b7ed348458649e2bfdf6a458944f781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-18 19:00:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efb43b7fcd837c1462ef6869dc58f592342e2769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:13:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e41ddb26aa43bc8413fa13c177d7c3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e7e11b6cf257469f9da71fe09cc11926 : SUCCESS in 33m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e006c737453e4bcb87a70f5e8eb0429d : SUCCESS in 1h 04m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cfbd2f33b9646d8b3feab37c7bb7f0e : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ecdc47a7bc8446daba908e94b8065ab : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91c37123282840d99549515908c54f33 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eba995da07c04459aeb1fab46909cdbc : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c97953279af42eaba22d0e8397f0889 : SUCCESS in 9m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/96a90335f3ef4f6babccf08562a2762b : SUCCESS in 3m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b67c9f9d1b514a0d81774cbc0fee58c2 : SUCCESS in 5m 32s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f7848c1e7ae34b9697b8822672a27259 : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"69bd795b889c277b17b8cd3017b2cb509df69988","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-25 11:50:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e73024a95e16b596bb6d7ef0eb89e5ac1bedeee","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-25 12:14:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f2d1b32cd422688a547962cc203e1e901a3f954","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-25 13:09:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f98e8826ce35d52506476185988166dd1d54154a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-25 13:11:22.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1a65328c25dc3212f4871725d69207f532795f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:21:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e2745a468e74257b7f5d5fad6c4efca\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/611f8620663947a18a05457cf4b51203 : SUCCESS in 1h 03m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/722aa94461df47f8880fdc518a2b7d18 : SUCCESS in 48m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bdfaefbad0d4946b0dc68957120015f : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e46f0fd7c6a14b8ab54498b7d99fa7d9 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4e31850c02842d5a3eb00ec7507f332 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66eab12d74c943e3ab17cd488c87ca4f : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3f0aea64c544158204eeba775ab11b : SUCCESS in 9m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be24bae4a5b249a381ac9d73dc61454f : SUCCESS in 6m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/983d83d0641447bc8015a91e070e15ba : SUCCESS in 9m 38s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6517a52ceb404062bd56952d88a195d7 : SUCCESS in 29m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bab5a3993be7035f5592f7efa85b1a23370d7f5c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-25 15:08:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dc25c2c2a66cb4faa1b1e164017544ed41f9879d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 15:09:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b38ec64ae904395bb32c03e53ac87a851ea5b3bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 16:14:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b6e4de312fb4574a707203ef102d9fa\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ee068e9094940208de24a4f2b6d3e49 : SUCCESS in 47m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ab7bb8932604d07a0a3d89d842efa30 : SUCCESS in 1h 01m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb283478cfe54b939335f3d728d37219 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9093a94706314ceabaf0f79b360fc91b : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47919af959924bc58e63973f0aee8451 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04110f8a752f47da8e9aea33a682a2cd : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00fcd5e53dbd485d838a23f7d67acd4f : SUCCESS in 6m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d72c10e77ea54f0bb630f96cc7675837 : SUCCESS in 7m 44s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cadcd0ee673c46ed94a0ac6350e17903 : SUCCESS in 25m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ab22cf8331ac5de90f06d4985b89fa51e58fa9d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 16:14:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f9d36a2955bda720a953d7030988991e51a90ed4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 16:15:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6420714c37034d238be1ecb043b328ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5e60e8f98a462d9ad77a32936f3459 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d57a957d520f0d74f6f9adf3e16fffbf27e7ec4a","revisions":{"079e6684b3572eea1e07651cb9c642db4cfb3c69":{"kind":"REWORK","_number":1,"created":"2025-11-18 19:00:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/02/967602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/02/967602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/02/967602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/02/967602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/02/967602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/02/967602/1"}}},"commit":{"parents":[{"commit":"99039e66c898a1a33c0f1b3cc9a220925adff984","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/99039e66c898a1a33c0f1b3cc9a220925adff984"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:58:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:59:31.000000000","tz":0},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nChange-Id: Ib0d97b69f513439e62a6e734ba4e278510a5b670\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/079e6684b3572eea1e07651cb9c642db4cfb3c69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/079e6684b3572eea1e07651cb9c642db4cfb3c69"}]},"branch":"refs/heads/master"},"d57a957d520f0d74f6f9adf3e16fffbf27e7ec4a":{"kind":"REWORK","_number":2,"created":"2025-11-25 13:11:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/02/967602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/02/967602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/02/967602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/02/967602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/02/967602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/02/967602/2"}}},"commit":{"parents":[{"commit":"99039e66c898a1a33c0f1b3cc9a220925adff984","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/99039e66c898a1a33c0f1b3cc9a220925adff984"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:58:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-25 13:11:22.000000000","tz":0},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nChange-Id: Ib0d97b69f513439e62a6e734ba4e278510a5b670\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d57a957d520f0d74f6f9adf3e16fffbf27e7ec4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d57a957d520f0d74f6f9adf3e16fffbf27e7ec4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
