)]}'
{"id":"openstack%2Fplacement~976906","triplet_id":"openstack%2Fplacement~master~I35fd8c5c6d1ace35cf968f4f309635366950ff0d","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 23:34:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35fd8c5c6d1ace35cf968f4f309635366950ff0d","subject":"Drop unused bandit target","status":"MERGED","created":"2026-02-16 08:56:33.000000000","updated":"2026-09-15 23:35:06.000000000","submitted":"2026-09-15 23:34:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976906","meta_rev_id":"e1dc48e8b548219d275ee173ba20d60e3416bc01","_number":976906,"virtual_id_number":976906,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 23:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-15 21:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-15 21:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 10:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-05 17:00:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-09-15 21:12:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8a49478100d1838a0573c19b2400cfaf306ed9ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-16 08:56:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b04a360d023c22c758e73367d812d37ac80b74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 10:08:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4771c5925264fb49eb5ea90ac97e0c4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/03c1661ad5544829869e3f720c235ad0 : SUCCESS in 3m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6f5c30bec114d2c8e56a9c1c4646d1a : SUCCESS in 34m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/192d7bd1f9d14b08aa8e32813906fd87 : SUCCESS in 1h 06m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1fe111e13104bb3a307b8a3c4964a25 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc932db7d43e4e6ca4ff43967deb4ef7 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c55ee2aee0b9479099e90da7ead521ac : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab816b2adce1415aaa7b3e64713c4a62 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32e4a9e2b6a417fa962e8615d4939e8 : FAILURE in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59f0519dbbd846e5bebe9f7a99d2e732 : SUCCESS in 3m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11a84829ebda462790f2f4866cd09508 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a0d64f116ccc4490bd778417616626b2 : SUCCESS in 4m 52s\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fec55bb5e769496b84dee529c72811b3 : SUCCESS in 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"acaeefb8e6d2b071cf82ec61a8006e6be60e12c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-05 17:00:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"4f3ceecec16586925739358bba8e7d1ffafa608d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:12:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/832feb345e274a8b8e8c7f0b76261da9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6a017c2dd1be4e94bd9acd8b67b5b080 : SUCCESS in 5m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6438e703f9c4d669a27ccff79016a4e : SUCCESS in 1h 05m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95175b053c6e4a0b8781622bf548cdaf : SUCCESS in 35m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd2639128fd421aad1eb8a59049fb72 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481e0c09cca443bc854b1af4129529d2 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c13775f0dbac4751a6dd3f36e9ec8a2d : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b942550cb8c845de9e549fc140ddd5ff : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96dd844b886745cbb580e9273ec3a10d : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a83b2eff4d045cd8a06f8d0cf5c7f54 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ed74244de3746c299702e445f6b7e3f : SUCCESS in 10m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ef9c79f6ee14b05b156dd869c87e091 : FAILURE in 4m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7fa2be75e83b42a49fc939fb240ef7de : SUCCESS in 4m 38s\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c74d22c3774f4efc8cca4e6b7aa3c708 : SUCCESS in 24m 42s","accounts_in_message":[],"_revision_number":2},{"id":"54d77c51fae212e2e62d834d972b028583edb92d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:12:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4b66649102a0831105c695d00309bf6e7a69f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:52:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d81c122aebc74f28b73f2be042941696\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/90ff8cf6b9004e5e971cf783520fbd41 : SUCCESS in 5m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38167072da0a44068fb298f231a7a629 : SUCCESS in 35m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0496b5a59df6472e950c139d44dbe29a : SUCCESS in 30m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccdf87729264465aa768f39ee7054094 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b2c9b300934860ac5c0419a190ae2e : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28f62573b37f4b6fbd867778c41a27ee : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/853ea781fd344632b304d1d39b3bcd00 : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c154b96adef4ad99603960afcca9dac : FAILURE in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dae1e3ea79847ceb4a83368fb26ca60 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e637389787e4047a0b6c6cc3e7358c0 : SUCCESS in 4m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1b9f59b96f174ab9a5ce73c0c235f9b6 : SUCCESS in 7m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/43fbcf8c2ec940518977343294c9e708 : SUCCESS in 3m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/67d61c321bd74c6b810e49d0dc695ea7 : SUCCESS in 10m 02s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a4adac0bd26d44fe8300cb689598c129 : SUCCESS in 22m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d0c61f4e35b7e66e5aef44579dbc86f95e35af0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:53:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b267cde6d6cc4c2550846145f792f870ab527e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 23:34:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62ab67eed5684da6b20a7cebaae6d97a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/865876f6d9c342a2a830bb86efe46935 : SUCCESS in 3m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8b91fe944aa4e14a6b095b5521dedab : SUCCESS in 1h 01m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4a79e89515e4ceeb7aa7816cbb2ef45 : SUCCESS in 57m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8c04e12db74489b40c11300af09079 : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac28aa02065b474ea2311acba637ce61 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d24f57eca704c0fbe52a5796f96b373 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43bb38c34ed84f3fba3d5de12f194fe1 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8c55af463cc424a95da7b9f4c6b5044 : SUCCESS in 9m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e981c0c3d3f947d5be401abc571873e3 : SUCCESS in 4m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ed6a017f2fb44553a3f301ca5ed99baf : SUCCESS in 5m 40s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/132d069aed6e4a548cde2fe2a72acad8 : SUCCESS in 5m 58s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bbd71b408f5b4070bfcdbbd7a870f598 : SUCCESS in 22m 23s","accounts_in_message":[],"_revision_number":2},{"id":"147353ae40025ea552d1e790eb130194f7c0780c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 23:34:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e1dc48e8b548219d275ee173ba20d60e3416bc01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 23:35:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a881fe93743c43e995fdb00f3f7cda27\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/79d26d663e8f4d4ebc1fbce844771fd1 : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708bb3cdb8fb4a838b551a18ccb2d49f : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6194221c88340bd87a23186506f97e0 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd3f4ee5368e0bc8b4c25207b02c476f821fd7a2","revisions":{"1defd7d2e0a6bdbb244b8ce20f3f7818ab541528":{"kind":"REWORK","_number":1,"created":"2026-02-16 08:56:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/976906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/06/976906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/976906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/976906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/976906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/06/976906/1"}}},"commit":{"parents":[{"commit":"6fdd8ed5c525b620cb9e696b6c5175095b9c62e6","subject":"Replace obsolete PCRE packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6fdd8ed5c525b620cb9e696b6c5175095b9c62e6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-16 08:55:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-16 08:55:28.000000000","tz":540},"subject":"Drop unused bandit target","message":"Drop unused bandit target\n\nThe bandit project was migrated to PyCQA[1] and the job was removed\nduring that migration.\n\n[1] https://review.opendev.org/c/openstack/bandit/+/564453\n\nChange-Id: I35fd8c5c6d1ace35cf968f4f309635366950ff0d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1defd7d2e0a6bdbb244b8ce20f3f7818ab541528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1defd7d2e0a6bdbb244b8ce20f3f7818ab541528"}]},"branch":"refs/heads/master"},"fd3f4ee5368e0bc8b4c25207b02c476f821fd7a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-05 17:00:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/06/976906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/06/976906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/976906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/976906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/976906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/06/976906/2"}}},"commit":{"parents":[{"commit":"d2727844011a8f144c818556ed7e8d43f756576f","subject":"Ubuntu: Use systemctl to restart services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d2727844011a8f144c818556ed7e8d43f756576f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-16 08:55:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-05 17:00:11.000000000","tz":0},"subject":"Drop unused bandit target","message":"Drop unused bandit target\n\nThe bandit project was migrated to PyCQA[1] and the job was removed\nduring that migration.\n\n[1] https://review.opendev.org/c/openstack/bandit/+/564453\n\nChange-Id: I35fd8c5c6d1ace35cf968f4f309635366950ff0d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fd3f4ee5368e0bc8b4c25207b02c476f821fd7a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fd3f4ee5368e0bc8b4c25207b02c476f821fd7a2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6fdd8ed5c525b620cb9e696b6c5175095b9c62e6","ours":"1defd7d2e0a6bdbb244b8ce20f3f7818ab541528","theirs":"d2727844011a8f144c818556ed7e8d43f756576f","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
