)]}'
{"id":"openstack%2Fplacement~996893","triplet_id":"openstack%2Fplacement~master~I6011fac25d1f24896ffa53a8d0aaaf534dd2de55","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-09 16:37:21.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-09 13:18:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-09 16:37:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6011fac25d1f24896ffa53a8d0aaaf534dd2de55","subject":"tox: Use constraints option","status":"MERGED","created":"2026-07-11 17:12:50.000000000","updated":"2026-09-09 16:39:42.000000000","submitted":"2026-09-09 16:37:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996893","meta_rev_id":"b8081199fd1c3037c8ff649cf564447fc10f7334","_number":996893,"virtual_id_number":996893,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 16:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-09-09 13:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-09-09 13:18:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-09-09 13:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-11 18:33:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 13:40:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-07 13:40:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"48876e336f8656e3276168ce758e9b0e01573d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adbf3cb7c84a4d6d6017bc92c20961f4505e73ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 18:33:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0692900d5024974b8e198d0a126ba94\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5901c283c2a341aeb7591afa9238cb9f : SUCCESS in 4m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ddfb1e7092494c05bdea5d9c6bc81f38 : SUCCESS in 1h 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64026ff1967c4b7abf2373823871b4cb : SUCCESS in 1h 01m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcec0046d0884bbc97ebf6174509258c : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4677cdb63aa04b4f95502551eb0bda30 : SUCCESS in 7m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fed0e20e301e4ad1b88a0c3a0f91b2cc : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ea26ef029fb4dfa9233be3cf59c356d : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a4327cbef66f4075b0d7886e0b022cef : FAILURE in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80182d572fc14bfaa2b986030c4e4953 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf833083ee594bc1a6d60db627572422 : SUCCESS in 8m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d361d13bc7eb430fb03acf552b130aa6 : SUCCESS in 5m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/450dd8241efa4f46b145537de055e43b : SUCCESS in 5m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4f4d1e130bf44633ba261fda1b53857c : SUCCESS in 8m 30s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cb4f6414e25a439e823a5d3adb7c2814 : SUCCESS in 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b3c816561f67b4e421fc761ee852b58a8c4b97ff","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:40:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d0bf139100fbf2b5bbff2df171eb561867ba0d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:49:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"eac73459c0c07aba33e14b213e69ee8945c1dcec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:50:45.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e152cfbe8c3d6ae8854881670b5037e4273e5e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:58:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d06c1b5f33d484893432e9da373fb75\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/56949fb11ec1479ab5ed0f54189777d5 : SUCCESS in 4m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e24a48bc275846309e2ed372e58e3977 : SUCCESS in 50m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5d7d428acc744aba568cc3053601383 : SUCCESS in 1h 00m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3ac84f29c5e43d8a5ba33ae00f78f54 : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7840ddb976643fc8ed973410c1da05e : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c214749ec1a645d882a64c46c57db2d7 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c635dbfa2754bec94e7a83fafdf23d1 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e0e5f12d743e46209533ecb4b8126ddb : FAILURE in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97da5ec8d744ec2a936843e1772338a : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d583b1fd3471453db98e6599ed7d941b : SUCCESS in 8m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c690c3fa0eee4e70923c916b5f3aa807 : SUCCESS in 5m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/379a45b45dc04f05a76bba4d7c923a81 : SUCCESS in 4m 47s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5f313a4fba6640dfbe9fd8072180e330 : SUCCESS in 7m 32s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/85f38375c7de4ff78483529cf83b5f13 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":3},{"id":"de58ba274bb927ed6ea3ce34c36d52abd95d3b04","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 16:48:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab9a78d2f027f7eb75db477d5923f60fbda6df4f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 16:48:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4edbc1cec7ac1499c44e00070811164c42757b2a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 16:48:43.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"03a52540391bc18c530cc334d5a42fde23f7b143","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-08 17:00:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0f6b9dec28e6686fdf232c28454d01824d049903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 13:15:52.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"0b401a5716cb191408f707f0f239af83f6a4ff2a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 13:18:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"534f4f340d337c43cb0322e8e10ddeb98fd29b03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:42:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84cda1be0b1243bda5d77708375119b7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e287641f6a3c40419c262112e821cea2 : SUCCESS in 5m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4bb22dab76d74e99b18ea400385c0f03 : SUCCESS in 1h 15m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87445da12dcb4a8c9d0a8d243012c74f : SUCCESS in 40m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cce1faaa58b43b98ef5777525a25ec1 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f1373237cf4f19ae98a8a7dc9ea872 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18666282793a4cbaa29d5d46761215eb : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84a7d8e00bf5471d969bbc456c9505ac : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/64e5d763dba043ee91502a7e66919312 : FAILURE in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5949cc4ba146bfb05cb4bd256259da : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de6d5180dfdb4aa8a859b2e536a07dad : SUCCESS in 7m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/31b79c38e6394ae2a859017d6ea86e0e : SUCCESS in 5m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6eeff7f14271412c9ecd06a6efff8683 : SUCCESS in 7m 47s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/318a9f1629974cc0923219f84c7cd355 : SUCCESS in 9m 00s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a52cbff67bd5442696686fb46c7b4f65 : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":4},{"id":"367b044c9c0f8fc41f77f1f3632ebb5c00cb958f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:43:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3de7bdf83226c0281ad9c9e3f2de592d6db3d22d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:37:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293ebec6c530470c9bc62d07d78b84f8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b46239987b9645c0a62d140d41f96687 : SUCCESS in 5m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41641e47043c46e8be4606d34b5a493d : SUCCESS in 1h 16m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c62907d4a1554d6a92f1c9f0218640af : SUCCESS in 42m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d3b9a0ac70485eb7aa47755b444e5e : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fe66056e3a3493f96324129d4482fc5 : SUCCESS in 7m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6938d2e8f474741b2aea9a362625989 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f6787e043f4370a1a38d7da6412f65 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a77732f45ac44698ecc26fb8f47f0cb : SUCCESS in 9m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f2056179c031486baf26e97bb23d9799 : SUCCESS in 6m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fefdc70f74904b61b390c134a9fc3530 : SUCCESS in 6m 02s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/72d6937e089b4d49b83cf7675459874e : SUCCESS in 9m 20s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ec032bba85424badbc30cf62836bfa78 : SUCCESS in 24m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d49c5224424c72a9c674ef986748880382745d0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:37:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b8081199fd1c3037c8ff649cf564447fc10f7334","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:39:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f26732dd34054fa68034d6ef3ac4fccd\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2cb77b806eff412084133d0e9b286d34 : SUCCESS in 1m 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9db4138fac34fe197eaf021022cc329 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce5b645226ba4d6faf03c376eca065fe : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"109ea4af9561aba4e67df8e6075055b764993489","revisions":{"9a982ea81a44e058938554ed760e153dfef49e27":{"kind":"REWORK","_number":1,"created":"2026-07-11 17:12:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/996893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/93/996893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/93/996893/1"}}},"commit":{"parents":[{"commit":"e3db398fba279721121892323e6260c6932797c1","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e3db398fba279721121892323e6260c6932797c1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:11:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:12:37.000000000","tz":540},"subject":"tox: Use constraints option","message":"tox: Use constraints option\n\nThis option was introduced in tox 4.28.0 to apply the constraints file\nconsistently.\n\nAlso remove the redundant override of the VIRTUAL_ENV environment.\nThe same value is already set by tox automatically.\n\nChange-Id: I6011fac25d1f24896ffa53a8d0aaaf534dd2de55\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9a982ea81a44e058938554ed760e153dfef49e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9a982ea81a44e058938554ed760e153dfef49e27"}]},"branch":"refs/heads/master"},"51c5874d13e3c7b1d7b6733d5994b6e949ee5d26":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-07 13:49:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/996893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/93/996893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/93/996893/2"}}},"commit":{"parents":[{"commit":"af1495d07d81de22cb0f9a663d8f97cff86b81dc","subject":"Replace license classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/af1495d07d81de22cb0f9a663d8f97cff86b81dc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:11:43.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:42:15.000000000","tz":60},"subject":"tox: Use constraints option","message":"tox: Use constraints option\n\nThis option was introduced in tox 4.28.0 to apply the constraints file\nconsistently.\n\nAlso remove the redundant override of the VIRTUAL_ENV environment.\nThe same value is already set by tox automatically.\n\nChange-Id: I6011fac25d1f24896ffa53a8d0aaaf534dd2de55\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/51c5874d13e3c7b1d7b6733d5994b6e949ee5d26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/51c5874d13e3c7b1d7b6733d5994b6e949ee5d26"}]},"branch":"refs/heads/master"},"da17c2795d4e70438ca4c3a899cdf434ebddc54d":{"kind":"NO_CHANGE","_number":3,"created":"2026-09-07 13:50:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/996893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/93/996893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/93/996893/3"}}},"commit":{"parents":[{"commit":"d87a53bf91d48219803e7b512b4715274b4bd342","subject":"Replace license classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d87a53bf91d48219803e7b512b4715274b4bd342"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:11:43.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-07-11 17:11:43.000000000","tz":540},"subject":"tox: Use constraints option","message":"tox: Use constraints option\n\nThis option was introduced in tox 4.28.0 to apply the constraints file\nconsistently.\n\nAlso remove the redundant override of the VIRTUAL_ENV environment.\nThe same value is already set by tox automatically.\n\nChange-Id: I6011fac25d1f24896ffa53a8d0aaaf534dd2de55\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/da17c2795d4e70438ca4c3a899cdf434ebddc54d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/da17c2795d4e70438ca4c3a899cdf434ebddc54d"}]},"branch":"refs/heads/master"},"109ea4af9561aba4e67df8e6075055b764993489":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-09 13:15:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/996893/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/93/996893/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/93/996893/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/93/996893/4"}}},"commit":{"parents":[{"commit":"5263171be71d124cd8a2d2276787a0033f5ada56","subject":"Merge \"Drop reference to lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5263171be71d124cd8a2d2276787a0033f5ada56"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:11:43.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-09 13:13:44.000000000","tz":60},"subject":"tox: Use constraints option","message":"tox: Use constraints option\n\nThis option was introduced in tox 4.28.0 to apply the constraints file\nconsistently.\n\nAlso remove the redundant override of the VIRTUAL_ENV environment.\nThe same value is already set by tox automatically.\n\nChange-Id: I6011fac25d1f24896ffa53a8d0aaaf534dd2de55\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/109ea4af9561aba4e67df8e6075055b764993489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/109ea4af9561aba4e67df8e6075055b764993489"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
