)]}'
{"id":"openstack%2Fproject-config~1002832","triplet_id":"openstack%2Fproject-config~master~Ia9bc0ccd12f1e7b731fbe3988812ed330bb8d9d1","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2026-08-30 14:22:34.000000000","reason":"removed on reply"},"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-09-02 15:26:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9bc0ccd12f1e7b731fbe3988812ed330bb8d9d1","subject":"Add a few more wandertracks and inaugust repos","status":"MERGED","created":"2026-08-28 22:04:12.000000000","updated":"2026-09-02 15:42:20.000000000","submitted":"2026-09-02 15:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002832","meta_rev_id":"96bff1c9ff9767a5c1a831d35f2fb1a56f59095d","_number":1002832,"virtual_id_number":1002832,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 15:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2026-08-30 14:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-02 15:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-02 15:15:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 22:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-29 00:58:24.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2026-09-02 15:15:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"39a1cf0be75159cd8283834d9b68140d62a9f845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-08-28 22:04:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee4d1f4b7f0bb7098d1eca9c5edd162433618cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:09:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40cb02402645472b801e254e1704ce2f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/613a733d17ea48d4ab3c49a2cbf06be9 : SUCCESS in 3m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/23aa7447cae7453198f4e3ba3305c97c : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a90f21ea312b5c2de533177e3013f88f98c02ae5","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-08-29 00:58:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb2454dd2a1918eeef8daf63e63e84bae251768f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-08-29 17:51:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"500bd8f312a5ff229145f1311a7eb4a1906a7e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 18:00:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebfdf44af247423fa384051125a257a5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/556fa46c77b946e8a764a06e0fdf3978 : SUCCESS in 4m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a0a5da81a8e140bf94025fbddd0734a7 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9b6a9a3289c9b5473b2a244146c79d153bf793c5","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-08-30 14:22:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4dbe95a74f6e931e4be1635d3437b5f001d7434f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-09-02 15:15:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d9526757b5b1bca52589c52c0bf56b56c1873154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:16:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e7953ec9d51b65bd1ddddc9cfff793d13497b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:26:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e702539c8054c7a8a5e737a8d5bceba\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f0211d6fd3041b09270b94b12eb11d1 : SUCCESS in 4m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1952ea95ebe64c7e82b74f580ef16fb9 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"45e368c87631e122c0a0aabd3118723abb1c7539","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:26:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"96bff1c9ff9767a5c1a831d35f2fb1a56f59095d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:42:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd5dd7e260094e55815684db8675d920\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/ed3cf577e182465d976e4db82310f9c6 : SUCCESS in 15m 02s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/6e3c290a21c145d1808e71fd5685c4d9 : SUCCESS in 6m 39s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/b6caefc709e14626a847887568bbb2a4 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2410bdc9867fe9a3b876b848b003af104c96bccd","revisions":{"d981484874407b468b573eda45baa3113f72d7fa":{"kind":"REWORK","_number":1,"created":"2026-08-28 22:04:12.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/32/1002832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/1002832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/1002832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/1002832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/1002832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/1002832/1"}}},"commit":{"parents":[{"commit":"d263a47d4b3e09b852e0764bbb33fa80a4aec3a1","subject":"kolla-check-sources: Stage the reno for commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d263a47d4b3e09b852e0764bbb33fa80a4aec3a1"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-08-28 22:03:06.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-08-28 22:03:06.000000000","tz":0},"subject":"Add a few more wandertracks and inaugust repos","message":"Add a few more wandertracks and inaugust repos\n\nI\u0027ve been holding off on these to keep from bugging people. But\nI think I\u0027ve got enough different things that don\u0027t belong in an\nexisting repo to justify another project-config patch.\n\nChange-Id: Ia9bc0ccd12f1e7b731fbe3988812ed330bb8d9d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d981484874407b468b573eda45baa3113f72d7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d981484874407b468b573eda45baa3113f72d7fa"}]},"branch":"refs/heads/master"},"2410bdc9867fe9a3b876b848b003af104c96bccd":{"kind":"REWORK","_number":2,"created":"2026-08-29 17:51:19.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/32/1002832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/1002832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/1002832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/1002832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/1002832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/1002832/2"}}},"commit":{"parents":[{"commit":"d263a47d4b3e09b852e0764bbb33fa80a4aec3a1","subject":"kolla-check-sources: Stage the reno for commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d263a47d4b3e09b852e0764bbb33fa80a4aec3a1"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-08-28 22:03:06.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-08-29 17:51:14.000000000","tz":0},"subject":"Add a few more wandertracks and inaugust repos","message":"Add a few more wandertracks and inaugust repos\n\nI\u0027ve been holding off on these to keep from bugging people. But\nI think I\u0027ve got enough different things that don\u0027t belong in an\nexisting repo to justify another project-config patch.\n\nChange-Id: Ia9bc0ccd12f1e7b731fbe3988812ed330bb8d9d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2410bdc9867fe9a3b876b848b003af104c96bccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2410bdc9867fe9a3b876b848b003af104c96bccd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
