)]}'
{"id":"openstack%2Fproject-config~1007930","triplet_id":"openstack%2Fproject-config~master~I3ffa3802a1356bdd97746364ed0eaeba3fa10dd6","project":"openstack/project-config","branch":"master","topic":"node-exporter-grafana-dashboard","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-10-01 12:50:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ffa3802a1356bdd97746364ed0eaeba3fa10dd6","subject":"Add a prometheus node exporter dashboard to grafana","status":"MERGED","created":"2026-09-29 16:53:41.000000000","updated":"2026-10-01 12:55:06.000000000","submitted":"2026-10-01 12:50:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007930-node-exporter-grafana-dashboard","meta_rev_id":"eeedf29c7ab172989b00b71dd2d1f2819b36db6c","_number":1007930,"virtual_id_number":1007930,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-01 12:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2026-09-29 17:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2026-10-01 12:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2026-10-01 12:36:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-29 17:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-29 17:42:20.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2026-10-01 12:36:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"00a0aa0aac04f2fc1e38c62d4009b0b4c7511705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-09-29 16:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8516f59cd58b83a9d9799a7ce6ec715e4fdb9a2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-09-29 17:05:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84e54cc3af2405bbe6a1787311b5bff85517cddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 17:29:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da5a8fc44924f9dbaac252efa9cb2e6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a6ee2faf895647e489ef288ffcb8ce50 : SUCCESS in 20m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/46583d7e1ed84768a2f6642fbe9af6d0 : SUCCESS in 4m 56s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/d53a439f80e74e61be6196c6c39fb8ad : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e189f4914eae25d869f1092b371498a3269a31c4","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-09-29 17:42:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a3b8bf272875e896d6208330f2d123a7a83f9d8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-10-01 12:36:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e63fa3864aacfbe3c9e1e304ac05ca945310b7ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 12:36:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b2170d1855dc2a635ddd6f17be0d2a30ca21037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 12:50:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312d8dd36c6e41a3bdc75d42e9fdd134\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/de78cbd7d02544f1a4cc5dacc52efede : SUCCESS in 13m 26s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8a36e61f04bc47e88e2281fd71cecf07 : SUCCESS in 4m 33s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/1fb76a91056b44b8b2acab1047580a32 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dcaa2947eb43d1e5710f5c9f7e80ca52ab627c8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 12:50:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eeedf29c7ab172989b00b71dd2d1f2819b36db6c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 12:55:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a52afc94ca8844938ce5a32cecbc4545\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/a6fdefa1842e4dbfa8889b1e110377c1 : SUCCESS in 3m 49s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/8c6fd3c1925445059f5141b51d2a6637 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22ca6f5ce9ab47e17076cce241f72b76c55b40ff","revisions":{"22ca6f5ce9ab47e17076cce241f72b76c55b40ff":{"kind":"REWORK","_number":1,"created":"2026-09-29 16:53:41.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/30/1007930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/1007930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/1007930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/1007930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/1007930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/1007930/1"}}},"commit":{"parents":[{"commit":"e30913af077f441dc6a5bb186b570c8cf28be204","subject":"Temporarily remove release docs semaphores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e30913af077f441dc6a5bb186b570c8cf28be204"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-09-29 16:49:57.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-09-29 16:49:57.000000000","tz":-420},"subject":"Add a prometheus node exporter dashboard to grafana","message":"Add a prometheus node exporter dashboard to grafana\n\nThis dashboard is taken from:\n  https://grafana.com/grafana/dashboards/1860-node-exporter-full/\n  https://github.com/rfmoz/grafana-dashboards/blob/master/prometheus/node-exporter-full.json\n\nthen a few small edits are made. I changed the uid value from the\ntemplate variable ${ds_prometheus} to the uid value for our prometheus\ndatasource. I dropped the templating entry that defines this variable\n(as it is no longer used). I switched the timezone from browser to UTC\n(as all our system logging is in UTC). I changed the name of the\nprometheus requirement to opendev-prometheus to match our datasource\nname. This may not actually be necessary or desired.\n\nThere may be additional tweaking we wish to do based on how this goes,\nbut for a first pass I\u0027m hopeful this gets us close to something usable.\n\nChange-Id: I3ffa3802a1356bdd97746364ed0eaeba3fa10dd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22ca6f5ce9ab47e17076cce241f72b76c55b40ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22ca6f5ce9ab47e17076cce241f72b76c55b40ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
