)]}'
{"id":"openstack%2Fproject-config~1008292","triplet_id":"openstack%2Fproject-config~master~I1beee5ed2e81eee0f61c214ed5d29636150babcb","project":"openstack/project-config","branch":"master","topic":"indri-branching-changes","hashtags":[],"change_id":"I1beee5ed2e81eee0f61c214ed5d29636150babcb","subject":"copy-release-tools-scripts: Remove add_release_note_page.sh","status":"NEW","created":"2026-10-01 15:31:13.000000000","updated":"2026-10-01 15:56:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"503e45400ba920652a83138ac9c96892a1e06d2a","_number":1008292,"virtual_id_number":1008292,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-01 15:56:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-01 15:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddf440afbeb98a560916516325fb88d439e1159c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 15:31:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9386f61fbf30f1968c5faeb9d72abdcb776a5d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 15:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8375c7a0552f49069a17b545fb66b632\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/025bbcfb73bf4e81b630c9e8823a1c34 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a56ccc7c88c0956d6a0699a936ae6f18988d3cb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 15:49:06.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"503e45400ba920652a83138ac9c96892a1e06d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 15:56:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f6a7295478f4fdb8e030d3c6d5b60da\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fa134a6a07b54741a211e75f0d17fd8a : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1557382412c0db6f9225bf51c14865a7ec44fa0c","revisions":{"28b0655b3ca89e1441f17e0b4fe97508acc1cfad":{"kind":"REWORK","_number":1,"created":"2026-10-01 15:31:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/1008292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/1008292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/1008292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/1008292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/1008292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/1008292/1"}}},"commit":{"parents":[{"commit":"bbe1047d1863b54852f2a1ceb44fbb77aa3e6183","subject":"copy-release-tools-scripts: Move release notes generation to a function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbe1047d1863b54852f2a1ceb44fbb77aa3e6183"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 12:21:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 14:51:37.000000000","tz":60},"subject":"copy-release-tools-scripts: Remove add_release_note_page.sh","message":"copy-release-tools-scripts: Remove add_release_note_page.sh\n\nJust call the new function directly in make_branch.sh, like we do for\nmost other things.\n\nChange-Id: I1beee5ed2e81eee0f61c214ed5d29636150babcb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28b0655b3ca89e1441f17e0b4fe97508acc1cfad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28b0655b3ca89e1441f17e0b4fe97508acc1cfad"}]},"branch":"refs/heads/master"},"1557382412c0db6f9225bf51c14865a7ec44fa0c":{"kind":"NO_CHANGE","_number":2,"created":"2026-10-01 15:49:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/1008292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/1008292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/1008292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/1008292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/1008292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/1008292/2"}}},"commit":{"parents":[{"commit":"cd5b6b9bf52b164b4529408d1e4f236c55eb4889","subject":"copy-release-tools-scripts: Move release notes generation to a function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd5b6b9bf52b164b4529408d1e4f236c55eb4889"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 12:21:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 15:48:56.000000000","tz":60},"subject":"copy-release-tools-scripts: Remove add_release_note_page.sh","message":"copy-release-tools-scripts: Remove add_release_note_page.sh\n\nJust call the new function directly in make_branch.sh, like we do for\nmost other things.\n\nChange-Id: I1beee5ed2e81eee0f61c214ed5d29636150babcb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1557382412c0db6f9225bf51c14865a7ec44fa0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1557382412c0db6f9225bf51c14865a7ec44fa0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
