)]}'
{"id":"openstack%2Fproject-config~1008396","triplet_id":"openstack%2Fproject-config~master~I70b436aff8a5e2267af80b85872e2962b0d9dfbd","project":"openstack/project-config","branch":"master","attention_set":{"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2026-10-07 20:33:35.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-10-07 19:10:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-10-07 07:20:08.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-10-07 20:33:36.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I70b436aff8a5e2267af80b85872e2962b0d9dfbd","subject":"Introduce barbican-ui-core group","status":"NEW","created":"2026-10-02 04:29:03.000000000","updated":"2026-10-07 20:33:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"f91386666a99b09af648f116d6d338dcf39adb9d","_number":1008396,"virtual_id_number":1008396,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-04 06:37:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-02 04:38:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-02 04:39:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"},{"updated":"2026-10-02 04:39:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-10-02 04:39:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d9201b0928f24035a59f019fd5d381eaffa51d87","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 04:29:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"951f2c78d7638bedf99eb54945f40a7ea23a8547","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 04:30:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3610f2f692c9149e4cfa87dd15a09487dccc093c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 04:38:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaf4ce3177964c10b6e985e6bd5783f7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f729ede466574a5ebc4590b3a235abd5 : SUCCESS in 2m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d76ab29594d4d60b71583d6dfdd48d5 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"63da517570fddd747b115c26b9bda2f17c710fa5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 04:38:51.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fbb4ec80f95ef9debdfac21ebdfaaa40217f5561","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 04:39:10.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"0730e603275288ab3ee682cba5465c90133d4a87","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 04:45:04.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"e71b77742425d34ef016fc77cab5374fb28e3de6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 04:48:30.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"151129c0d6800b27177c7a4105fdbc0e7903a0a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 04:52:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0072445cf649463b8c1c633ee9be9549\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9fbc250ef9764ab4805b4f099004018a : SUCCESS in 2m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8b6c0f5708944f7296047cee3afd310e : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":4},{"id":"8acd38590df57305ea127f9c58158587b3c5edbe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 04:53:01.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"50ff8645bf8a672868d3ab8aee4707a436580c66","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-10-02 16:54:05.000000000","message":"Patch Set 4: Code-Review-1\n\nThis ACL won\u0027t be used unless you remove the acl-config option from barbican-ui in the gerrit/projects.yaml file.\n\nAlso this change adds job templates to zuul.d/projects.yaml which aren\u0027t mentioned or explained in your commit message.\n\nFurther, the new ACL deviates noticeably from barbican\u0027s ACL, dropping separate stable branch permissions and adding label-backport-candidate (along with a label definition) as well as toggleWipState yet none of these decisions seem to be explained in your commit message nor the meeting log linked from there.\n\nThe decision to separate out permissions for barbican-ui does seem to be recorded in the meeting, but we\u0027ll still need Mauricio to +1 this change as PTL due to the various other choices which have been made herein.","accounts_in_message":[],"_revision_number":4},{"id":"77bec7245bbe3424a1399c07a26ec15a14f2a751","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 05:42:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"764e6c4627a16733417c93db853812c41aceccf1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 05:49:31.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7b89f3eae4eaa5131307dd700c08e387608be46d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 05:50:11.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"bfa69f2e4f24f21a980ad14f6c68383d68e9b108","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 05:52:39.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"67cdee9bea74c27a031169d2401bb90e007b43b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 05:54:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5633a07ded454ac08d06fc9112a6c29f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/02e6e7e9151246228cc66cc204d418cf : SUCCESS in 3m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a3468ae037a84c0ea38da747b5c809f4 : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":6},{"id":"4bb027a04911c301ef639763375c289833c1c769","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 05:55:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"c11ba2ffe951ee96ab4349dc296a5282b3320343","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 06:01:14.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"a626da5a4844e2f7cc49a669b294a448de1c31d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 06:10:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/948952333a2d4c60a3506d2433db5313\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/63037a5d45ab43e1878fb79c293814a7 : SUCCESS in 4m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/253cfb5157014513bed99d460bfc7cfe : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":7},{"id":"6a3cdea0dc3696fcf5b1aa9f0c1b0010de28f8b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 06:12:51.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"b4a8f1d1127bf3c2a292eeb204c9bf11b118657a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 06:15:06.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"60ff1034505b83002ba1d6dba5532390a4b67f06","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 06:15:35.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"d7bdd2d77c0a37fa7459b1e6d88e3387419ea921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 06:24:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d06534c0937b45df91f843c77b3c58d7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/331a40abbb364153abcae41f64372fb3 : SUCCESS in 4m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f07cf3c3cfb64bea8b63ef58e9e38813 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":10},{"id":"2e9c828c1825809dba9f70de71318ce38eb8fa1a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 06:42:00.000000000","message":"Patch Set 10:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":10},{"id":"77df35f67d6c38406de50857ec62c76c107160e9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-10-03 12:34:06.000000000","message":"Patch Set 10:\n\nThis latest revision adds groups:barbican to the projects.yaml file, which will cause our integration tooling to expect barbican-ui bugs filed on the https://launchpad.net/barbican project instead of https://launchpad.net/barbican-ui . This may be intentional since the latter seems to have been created 10 years ago and never configured for bug tracking, but I wanted to double-check it\u0027s an intentional adjustment to the change as it\u0027s not mentioned in the commit message.\n\nYou\u0027re also still adding the publish-to-pypi job template, even though barbican-ui has never been published to pypi.org in the past or even tagged for release at all, so I just want to make sure it\u0027s intentional as there\u0027s still no mention of it in the commit message for this change. If you do plan to start releasing it, you\u0027ll also want to add related deliverable configuration in the openstack/releases repository.\n\nAs long as Mauricio gives a +1 though, we can go ahead and merge it at this point.","accounts_in_message":[],"_revision_number":10},{"id":"280bfe05ea5af254ec2140b21061cb7c4c4b8ae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-04 06:20:01.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"5ccd6bd8ed8db8a8e17fb7e9ad213ea5f3b295c2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-04 06:21:59.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"976e75cbfcdac1d6891bc40e87eaea01b6eec6cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-04 06:29:41.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/571ed3dac6ec4890a29a45818d78772f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/571d0763d86c4cc2a00fd9ea9560be9b : SUCCESS in 4m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/009b5ab877ab46aa97f2a054e1d27d92 : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":11},{"id":"2af7b55017ca1ad3e3e7aa89d21c29c38eb3dc60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-04 06:34:19.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"8fb1f4c819778644d4c23b2a48d7a99c980f7cc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-04 06:37:56.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1245325ae9d54e33aed510aed2937cac\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/953df9fc484e40708ba676709d88251e : SUCCESS in 2m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/aaddba4563ee4da4bf5892754dbe68d4 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":12},{"id":"d9f299a094dd0ed16f2c7d79a13c6c6ad05ec7e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-10-07 07:20:08.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5521168aed5410810db90513e9aa3fdfe235b924","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-10-07 18:25:41.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"aa5e4f27b6f47dd699fbdedddd3398674e8f8d54","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-10-07 19:10:45.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"9912a58e4535a8d41b8cbd0a554c4a42cc9406ee","revisions":{"24c61fcddb8d66b27d4cd7e589024584fd1bce3b":{"kind":"REWORK","_number":1,"created":"2026-10-02 04:29:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/1"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"subject":"WIP: Introduce barbican-ui-core group","message":"WIP: Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24c61fcddb8d66b27d4cd7e589024584fd1bce3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24c61fcddb8d66b27d4cd7e589024584fd1bce3b"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"7f5a4eac3ace4628dbe2de8e2bf95f8ba484b0d9":{"kind":"REWORK","_number":2,"created":"2026-10-02 04:30:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/2"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:30:57.000000000","tz":0},"subject":"WIP: Introduce barbican-ui-core group","message":"WIP: Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f5a4eac3ace4628dbe2de8e2bf95f8ba484b0d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f5a4eac3ace4628dbe2de8e2bf95f8ba484b0d9"}]},"branch":"refs/heads/master"},"143eecccb6d004963b1fbdf64d1d02eba1e1c0f4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-10-02 04:38:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/3"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:38:51.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143eecccb6d004963b1fbdf64d1d02eba1e1c0f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143eecccb6d004963b1fbdf64d1d02eba1e1c0f4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8c0e6cc6706f0d11248ca6a8fc1707405ef0b5b6":{"kind":"REWORK","_number":4,"created":"2026-10-02 04:48:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/4"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:46:32.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c0e6cc6706f0d11248ca6a8fc1707405ef0b5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c0e6cc6706f0d11248ca6a8fc1707405ef0b5b6"}]},"branch":"refs/heads/master"},"3cf67f422c1175e3e5e96adee293a94233c16ca4":{"kind":"REWORK","_number":5,"created":"2026-10-03 05:49:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/5"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-03 05:49:17.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3cf67f422c1175e3e5e96adee293a94233c16ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3cf67f422c1175e3e5e96adee293a94233c16ca4"}]},"branch":"refs/heads/master"},"c2716a607161ddfd7375136f31e1596c75808451":{"kind":"REWORK","_number":6,"created":"2026-10-03 05:50:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/6"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-03 05:50:11.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c2716a607161ddfd7375136f31e1596c75808451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c2716a607161ddfd7375136f31e1596c75808451"}]},"branch":"refs/heads/master"},"d011530421667a954ba292b39db9c80469190e9d":{"kind":"REWORK","_number":7,"created":"2026-10-03 06:01:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/7"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-03 06:01:13.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d011530421667a954ba292b39db9c80469190e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d011530421667a954ba292b39db9c80469190e9d"}]},"branch":"refs/heads/master"},"4d07017d0e9676508d5bdc13c803215e5d7fde94":{"kind":"REWORK","_number":8,"created":"2026-10-03 06:12:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/8"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-03 06:12:50.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d07017d0e9676508d5bdc13c803215e5d7fde94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d07017d0e9676508d5bdc13c803215e5d7fde94"}]},"branch":"refs/heads/master"},"8a28116e45c9d3fdd45cc280a5b27974820306ae":{"kind":"REWORK","_number":9,"created":"2026-10-03 06:15:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/9"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-03 06:15:04.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a28116e45c9d3fdd45cc280a5b27974820306ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a28116e45c9d3fdd45cc280a5b27974820306ae"}]},"branch":"refs/heads/master"},"c3ab8564e5b877a8654786ebe42a9112a677f182":{"kind":"REWORK","_number":10,"created":"2026-10-03 06:15:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/10"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-03 06:15:35.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3ab8564e5b877a8654786ebe42a9112a677f182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3ab8564e5b877a8654786ebe42a9112a677f182"}]},"branch":"refs/heads/master"},"46733b0506f8673aefb7291538ccfc201d5c939e":{"kind":"REWORK","_number":11,"created":"2026-10-04 06:20:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/11"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-04 06:19:36.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/46733b0506f8673aefb7291538ccfc201d5c939e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/46733b0506f8673aefb7291538ccfc201d5c939e"}]},"branch":"refs/heads/master"},"9912a58e4535a8d41b8cbd0a554c4a42cc9406ee":{"kind":"REWORK","_number":12,"created":"2026-10-04 06:34:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/96/1008396/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/1008396/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/1008396/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/1008396/12"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-02 04:29:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-10-04 06:34:19.000000000","tz":0},"subject":"Introduce barbican-ui-core group","message":"Introduce barbican-ui-core group\n\nA decision was made to organize a separate team to work on the Barbican-UI.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/barbican/2026/barbican.2026-09-28-15.00.log.html\n\nPlease also add in this group members, Barbican PTL:\nMauricio Harley - mharley@redhat.com\n\nAnd include in Included Groups:\nbarbican-core\n\nChange-Id: I70b436aff8a5e2267af80b85872e2962b0d9dfbd\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9912a58e4535a8d41b8cbd0a554c4a42cc9406ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9912a58e4535a8d41b8cbd0a554c4a42cc9406ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
