)]}'
{"id":"openstack%2Fproject-config~1008533","triplet_id":"openstack%2Fproject-config~master~Ieb34f2b7c2affc7a99954419c7989543dc03011d","project":"openstack/project-config","branch":"master","topic":"cleanup-afs-wheel-graphs","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-10-06 20:19:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieb34f2b7c2affc7a99954419c7989543dc03011d","subject":"Cleanup afs grafana dashboard wheel graphs","status":"MERGED","created":"2026-10-02 20:23:00.000000000","updated":"2026-10-06 20:25:19.000000000","submitted":"2026-10-06 20:19:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1008533-cleanup-afs-wheel-graphs","meta_rev_id":"76e2576b6bab3bd67f67a5c76c8964febf4cd2d7","_number":1008533,"virtual_id_number":1008533,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-06 20:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2026-10-06 19:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-10-06 19:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2026-10-06 19:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-02 20:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-06 19:49:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-10-06 19:54:24.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"b9501d94c8ef0feef7cda33918a95f550005df76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-10-02 20:23:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c44ac30a48f41f85a8bc417e9214f9b9c9bcf161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 20:53:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ee10c13a2f41f5b55e2e7954aa9f13\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/97eb1c02b2c94a46b753ea4b885b6a2a : SUCCESS in 25m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/70e85d0832d94e6e878e9605a8075e71 : SUCCESS in 5m 07s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/f3668f7accef41ad99de8c456f403412 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f4cae43ad856eb85ef94266ebcad2d3baa8960f9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-10-06 19:49:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3413e544ba8a02d2eff870df83fdecf3d573e941","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-10-06 19:54:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1efc9721e85272b7623d910e375637530a9aaaba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 19:54:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d03e227d5105a5483983041fb59572ec4144867a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 20:19:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6b6cb61ad2446f590be715afd8d31ff\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/47d00bcbd30042d1b1ee2ab172327f6b : SUCCESS in 20m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ff5e2163cd6241e4b4ec31bc3e755d4a : SUCCESS in 6m 02s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/53c5e317c00e4a52b9830a545183c240 : SUCCESS in 15m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2e4940baf7fed80779f15fd3ce73f8b9e98e54e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 20:19:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"76e2576b6bab3bd67f67a5c76c8964febf4cd2d7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 20:25:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93221e25c19f4acaa0ee0e76514228ec\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/70700d392b624bdf8981cb79cdcd0677 : SUCCESS in 3m 31s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/4ff6c7d0308c49a087d60bcb7f88bf7e : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c22a3451708b6eb1aa907b3aaf19f1f644c0c3e7","revisions":{"c22a3451708b6eb1aa907b3aaf19f1f644c0c3e7":{"kind":"REWORK","_number":1,"created":"2026-10-02 20:23:00.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/33/1008533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/1008533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/1008533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/1008533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/1008533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/1008533/1"}}},"commit":{"parents":[{"commit":"8d9478ccbad5b4de7d48da1bb5677048546d03f7","subject":"Merge \"Add a prometheus node exporter dashboard to grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d9478ccbad5b4de7d48da1bb5677048546d03f7"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-10-02 20:22:09.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-10-02 20:22:09.000000000","tz":-420},"subject":"Cleanup afs grafana dashboard wheel graphs","message":"Cleanup afs grafana dashboard wheel graphs\n\nI just deleted centos 7, debian buster, ubuntu xenial, and ubuntu bionic\nwheel volumes from afs. Update the grafana dashboard to match. While we\nare add it add in Jammy graphs as those were missing.\n\nChange-Id: Ieb34f2b7c2affc7a99954419c7989543dc03011d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c22a3451708b6eb1aa907b3aaf19f1f644c0c3e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c22a3451708b6eb1aa907b3aaf19f1f644c0c3e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
