)]}'
{"id":"openstack%2Fproject-config~152280","triplet_id":"openstack%2Fproject-config~master~I055cf6a7a0ae4737e24800dd6968359505663df5","project":"openstack/project-config","branch":"master","topic":"updown","hashtags":[],"change_id":"I055cf6a7a0ae4737e24800dd6968359505663df5","subject":"add devstack up/down test","status":"MERGED","created":"2015-02-02 21:30:21.000000000","updated":"2015-02-03 21:17:07.000000000","submitted":"2015-02-03 21:17:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"30efe6cb8e2ec72505a80fe5b81bece346c8a5bf","_number":152280,"virtual_id_number":152280,"owner":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2015-02-03 21:17:06.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2015-02-02 21:31:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2015-02-03 21:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-02-02 21:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2015-02-03 21:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-02-02 21:31:10.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2015-02-02 21:31:59.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2015-02-03 21:10:35.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2015-02-03 21:17:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c3fc9ad7db1a24a4c22b20abd8b8852d2e5762cb","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-02-02 21:30:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fa6f7d587cfa1132b5a855ff37c09183d494246","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2015-02-02 21:31:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"569e4d7dec2134daaabeec595a376291c6f7da3e","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2015-02-02 21:31:59.000000000","message":"Patch Set 1: Code-Review+1\n\ngood idea","accounts_in_message":[],"_revision_number":1},{"id":"1751411ead52574c54f5f838526d1a224f88023a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-02 21:37:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-infra-puppet-apply-precise http://logs.openstack.org/80/152280/1/check/gate-infra-puppet-apply-precise/b606efc/ : SUCCESS in 2m 52s\n- gate-infra-puppet-apply-trusty http://logs.openstack.org/80/152280/1/check/gate-infra-puppet-apply-trusty/1c2c83a/ : SUCCESS in 2m 28s\n- gate-infra-puppet-apply-centos6 http://logs.openstack.org/80/152280/1/check/gate-infra-puppet-apply-centos6/4487d90/ : SUCCESS in 6m 44s\n- project-config-compare-xml http://logs.openstack.org/80/152280/1/check/project-config-compare-xml/d7de0f6/ : Jenkins XML output has changed. in 4m 35s (non-voting)\n- project-config-gerrit http://logs.openstack.org/80/152280/1/check/project-config-gerrit/111a58e/ : SUCCESS in 33s\n- gate-project-config-layout http://logs.openstack.org/80/152280/1/check/gate-project-config-layout/dbba090/ : SUCCESS in 3m 31s\n- gate-project-config-pep8 http://logs.openstack.org/80/152280/1/check/gate-project-config-pep8/3e10b77/ : SUCCESS in 29s\n- gate-project-config-bashate http://logs.openstack.org/80/152280/1/check/gate-project-config-bashate/643692a/ : SUCCESS in 22s\n- gate-project-config-jenkins-project http://logs.openstack.org/80/152280/1/check/gate-project-config-jenkins-project/09dd664/ : SUCCESS in 18s\n- gate-project-config-projects-yaml http://logs.openstack.org/80/152280/1/check/gate-project-config-projects-yaml/4a2c9b9/ : SUCCESS in 20s\n- check-generate-specs-site http://logs.openstack.org/80/152280/1/check/check-generate-specs-site/275eb91/ : SUCCESS in 11s","accounts_in_message":[],"_revision_number":1},{"id":"d864ef042fda12dd3ba3f3274ab276bdf26ca7fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-03 21:10:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"9d1d01f9df744563b19aaafc5fc9098031276ae5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2015-02-03 21:10:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95fb0b47bf0bca0025c05ca109e88a1e69dc007a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-03 21:17:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-infra-puppet-apply-precise http://logs.openstack.org/80/152280/1/gate/gate-infra-puppet-apply-precise/523bad8/ : SUCCESS in 4m 35s\n- gate-infra-puppet-apply-trusty http://logs.openstack.org/80/152280/1/gate/gate-infra-puppet-apply-trusty/18f40c6/ : SUCCESS in 2m 19s\n- gate-infra-puppet-apply-centos6 http://logs.openstack.org/80/152280/1/gate/gate-infra-puppet-apply-centos6/f337ed7/ : SUCCESS in 6m 28s\n- gate-project-config-layout http://logs.openstack.org/80/152280/1/gate/gate-project-config-layout/9b5450e/ : SUCCESS in 2m 51s\n- gate-project-config-pep8 http://logs.openstack.org/80/152280/1/gate/gate-project-config-pep8/60c3958/ : SUCCESS in 28s\n- gate-project-config-bashate http://logs.openstack.org/80/152280/1/gate/gate-project-config-bashate/7aff0cd/ : SUCCESS in 19s\n- gate-project-config-jenkins-project http://logs.openstack.org/80/152280/1/gate/gate-project-config-jenkins-project/bb5c424/ : SUCCESS in 19s\n- gate-project-config-projects-yaml http://logs.openstack.org/80/152280/1/gate/gate-project-config-projects-yaml/7f138c4/ : SUCCESS in 27s\n- check-generate-specs-site http://logs.openstack.org/80/152280/1/gate/check-generate-specs-site/9ab8d94/ : SUCCESS in 18s","accounts_in_message":[],"_revision_number":1},{"id":"1678ba9be1640f8d623ed4f88a0180b6cb163000","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-03 21:17:07.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9a3d34a97215163fb8fe271bb1a8e8dd1dffafe","revisions":{"d9a3d34a97215163fb8fe271bb1a8e8dd1dffafe":{"kind":"REWORK","_number":1,"created":"2015-02-02 21:30:21.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/80/152280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/152280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/152280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/152280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/152280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/152280/1"}}},"commit":{"parents":[{"commit":"fbc4fb4fbde311086e784c146a6de2fdef5d9c10","subject":"Revert \"Revert \"Push dg-tempest-dsvm-full logs to swift\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fbc4fb4fbde311086e784c146a6de2fdef5d9c10"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2015-02-02 21:28:43.000000000","tz":-300},"committer":{"name":"Sean Dague","email":"sean@dague.net","date":"2015-02-02 21:28:43.000000000","tz":-300},"subject":"add devstack up/down test","message":"add devstack up/down test\n\nthis is a simple functional test for devstack that ensures that the\nreset scripts correctly shut things down at the end of the run. With\nthis in place we\u0027ll make the cleanup scripts do a set -o errexit.\n\nThis will help prevent us from breaking repeatability in devstack.\n\nChange-Id: I055cf6a7a0ae4737e24800dd6968359505663df5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9a3d34a97215163fb8fe271bb1a8e8dd1dffafe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9a3d34a97215163fb8fe271bb1a8e8dd1dffafe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
