)]}'
{"id":"openstack%2Fproject-config~278978","triplet_id":"openstack%2Fproject-config~master~I788cd70dc4e6a9c3f8e6805d5bdcdf7f549fce35","project":"openstack/project-config","branch":"master","topic":"rm_largeops","hashtags":[],"change_id":"I788cd70dc4e6a9c3f8e6805d5bdcdf7f549fce35","subject":"remove large-ops testing","status":"MERGED","created":"2016-02-11 11:01:21.000000000","updated":"2016-02-12 11:54:54.000000000","submitted":"2016-02-12 11:54:53.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1211413c829b6c49b04e89497c71174c582664e9","_number":278978,"virtual_id_number":278978,"owner":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-02-12 11:54:53.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-02-12 09:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2016-02-12 11:45:27.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2016-02-12 11:45:27.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-12 09:34:49.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2016-02-12 11:45:27.000000000","updated_by":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"reviewer":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"state":"REVIEWER"},{"updated":"2016-02-12 11:54:53.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7c306a53e461c1ba51a414cc0aeabff5cd1a7d71","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-02-11 11:01:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d6198944fbc2adeab7a0fc1d7723ebab75c3e70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-11 11:50:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-project-config-compare-xml http://logs.openstack.org/78/278978/1/check/gate-project-config-compare-xml/c0f02c8/ : Jenkins XML output has changed. in 6m 22s (non-voting)\n- gate-project-config-layout http://logs.openstack.org/78/278978/1/check/gate-project-config-layout/e9bddba/ : FAILURE in 50s","accounts_in_message":[],"_revision_number":1},{"id":"f3ffdf9ca9f42140437c9cf607434311a31ebbb0","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-02-11 11:55:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8280dd1f44f0b76d93e8096796e9896b34df3ed2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2016-02-11 12:40:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fb3ddda1ccf1040fa6ae5817ea381ac4b66c3b77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-11 13:00:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-compare-xml http://logs.openstack.org/78/278978/2/check/gate-project-config-compare-xml/10e56e3/ : Jenkins XML output has changed. in 10m 02s (non-voting)\n- gate-project-config-layout http://logs.openstack.org/78/278978/2/check/gate-project-config-layout/65df894/ : SUCCESS in 6m 00s\n- gate-project-config-jenkins-project http://logs.openstack.org/78/278978/2/check/gate-project-config-jenkins-project/cdeab42/ : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"3b88e4d17f2f998bc81bb224be56b4f6e85ff2dd","author":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"date":"2016-02-12 09:11:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8bb7d65b5fd2e8d5948e54f92f4a2996afc136b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 09:12:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"598bea14b4e21533e5a2bb59caa5b55cb93a53e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 09:18:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/78/278978/2/gate/gate-project-config-layout/b59c19e/ : SUCCESS in 6m 24s\n- gate-project-config-jenkins-project http://logs.openstack.org/78/278978/2/gate/gate-project-config-jenkins-project/d70647d/ : SUCCESS in 30s","accounts_in_message":[],"_revision_number":2},{"id":"d224a9786946cbb4d875d40058d736a0ee29bc36","date":"2016-02-12 09:18:49.000000000","message":"Change could not be merged due to a path conflict.\n\nPlease rebase the change locally and upload the rebased commit for review.","accounts_in_message":[],"_revision_number":2},{"id":"00df28f9a2338495549d08ddbbda9023515e3b20","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2016-02-12 09:34:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c368b3c458e517f4aab18ccdb82e2ff072615739","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2016-02-12 09:34:49.000000000","message":"Patch Set 3: Code-Review+2\n\nrebased","accounts_in_message":[],"_revision_number":3},{"id":"1d9725deb9bfd1a1efc5008019840b456b36ed0d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 09:50:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-compare-xml http://logs.openstack.org/78/278978/3/check/gate-project-config-compare-xml/e4d1fc0/ : Jenkins XML output has changed. in 15m 06s (non-voting)\n- gate-project-config-layout http://logs.openstack.org/78/278978/3/check/gate-project-config-layout/be6d2b4/ : SUCCESS in 7m 09s\n- gate-project-config-jenkins-project http://logs.openstack.org/78/278978/3/check/gate-project-config-jenkins-project/f86ba72/ : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3},{"id":"4e2428c76a497c825f1d95ce5e8e8e5dcfe32a11","author":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"date":"2016-02-12 11:45:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e14c5ee22080f5bf2f86c4dead83cde735fab447","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 11:45:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"635066f19e344dcf8b11bc84b55342d554afe370","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 11:54:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/78/278978/3/gate/gate-project-config-layout/b19ac3a/ : SUCCESS in 9m 04s\n- gate-project-config-jenkins-project http://logs.openstack.org/78/278978/3/gate/gate-project-config-jenkins-project/adfba5b/ : SUCCESS in 21s","accounts_in_message":[],"_revision_number":3},{"id":"4df21617d44b572522ab159082e07ad6e91417c0","date":"2016-02-12 11:54:54.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"36dc536e15705d8c7baeaf5f61d8a6f6bfb861ef","revisions":{"314bc1ce5125031fcf3cca88115ef11cfe266b10":{"kind":"REWORK","_number":1,"created":"2016-02-11 11:01:21.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/78/278978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/278978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/278978/1"}}},"commit":{"parents":[{"commit":"db6e21ac9b58acc2d32454bde30b77eb3eb9c909","subject":"Merge \"Remove gate-nova-pylint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db6e21ac9b58acc2d32454bde30b77eb3eb9c909"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-02-11 10:59:42.000000000","tz":-300},"committer":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-02-11 10:59:42.000000000","tz":-300},"subject":"remove large-ops testing","message":"remove large-ops testing\n\nPer email thread there is general support to drop these tests which\naren\u0027t really getting debugged, and failing good patches, and work\ntowards a better approach here using rally / performance counters.\n\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-February/086232.html\n\nChange-Id: I788cd70dc4e6a9c3f8e6805d5bdcdf7f549fce35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/314bc1ce5125031fcf3cca88115ef11cfe266b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/314bc1ce5125031fcf3cca88115ef11cfe266b10"}]},"branch":"refs/heads/master"},"e23f1dbd0ecd6c314365bc9cad25892bc3e7bbcc":{"kind":"REWORK","_number":2,"created":"2016-02-11 11:55:53.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/78/278978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/278978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/278978/2"}}},"commit":{"parents":[{"commit":"db6e21ac9b58acc2d32454bde30b77eb3eb9c909","subject":"Merge \"Remove gate-nova-pylint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db6e21ac9b58acc2d32454bde30b77eb3eb9c909"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-02-11 10:59:42.000000000","tz":-300},"committer":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-02-11 11:55:47.000000000","tz":-300},"subject":"remove large-ops testing","message":"remove large-ops testing\n\nPer email thread there is general support to drop these tests which\naren\u0027t really getting debugged, and failing good patches, and work\ntowards a better approach here using rally / performance counters.\n\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-February/086232.html\n\nChange-Id: I788cd70dc4e6a9c3f8e6805d5bdcdf7f549fce35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e23f1dbd0ecd6c314365bc9cad25892bc3e7bbcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e23f1dbd0ecd6c314365bc9cad25892bc3e7bbcc"}]},"branch":"refs/heads/master"},"36dc536e15705d8c7baeaf5f61d8a6f6bfb861ef":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-02-12 09:34:20.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/78/278978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/278978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/278978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/278978/3"}}},"commit":{"parents":[{"commit":"252d4e3e1911a97fa368e24672dc3a0d6c5b5fb1","subject":"Merge \"import release-test repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/252d4e3e1911a97fa368e24672dc3a0d6c5b5fb1"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-02-11 10:59:42.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2016-02-12 09:33:17.000000000","tz":60},"subject":"remove large-ops testing","message":"remove large-ops testing\n\nPer email thread there is general support to drop these tests which\naren\u0027t really getting debugged, and failing good patches, and work\ntowards a better approach here using rally / performance counters.\n\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-February/086232.html\n\nChange-Id: I788cd70dc4e6a9c3f8e6805d5bdcdf7f549fce35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36dc536e15705d8c7baeaf5f61d8a6f6bfb861ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36dc536e15705d8c7baeaf5f61d8a6f6bfb861ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
