)]}'
{"id":"openstack%2Fproject-config~302072","triplet_id":"openstack%2Fproject-config~master~I57c0b06d5d4444dc4c1afbb897612097dac92718","project":"openstack/project-config","branch":"master","topic":"smaug-python3","hashtags":[],"change_id":"I57c0b06d5d4444dc4c1afbb897612097dac92718","subject":"Smaug: add python3 CI jobs","status":"MERGED","created":"2016-04-06 07:44:33.000000000","updated":"2016-04-06 12:06:41.000000000","submitted":"2016-04-06 12:06:41.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"87fb97467b1b20f141cce87ce1f0ea1c50beab38","_number":302072,"virtual_id_number":302072,"owner":{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2016-04-06 12:06:40.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17192,"name":"xiangxinyong","email":"xiangxinyong@huawei.com","username":"xianxinyong"},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":0,"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2016-04-06 11:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-04-06 11:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17192,"name":"xiangxinyong","email":"xiangxinyong@huawei.com","username":"xianxinyong"},{"value":2,"date":"2016-04-06 10:24:56.000000000","_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":1,"date":"2016-04-06 08:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2016-04-06 11:54:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17192,"name":"xiangxinyong","email":"xiangxinyong@huawei.com","username":"xianxinyong"},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":0,"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},{"_account_id":17192,"name":"xiangxinyong","email":"xiangxinyong@huawei.com","username":"xianxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-04-06 08:46:45.000000000","updated_by":{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},"reviewer":{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},"state":"REVIEWER"},{"updated":"2016-04-06 10:24:56.000000000","updated_by":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"reviewer":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"state":"REVIEWER"},{"updated":"2016-04-06 11:21:32.000000000","updated_by":{"_account_id":17192,"name":"xiangxinyong","email":"xiangxinyong@huawei.com","username":"xianxinyong"},"reviewer":{"_account_id":17192,"name":"xiangxinyong","email":"xiangxinyong@huawei.com","username":"xianxinyong"},"state":"REVIEWER"},{"updated":"2016-04-06 11:54:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2016-04-06 12:06:40.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"eb0fba3b0d65e3344e0035a99561f9c63a48178f","author":{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},"date":"2016-04-06 07:44:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8b45fd1712afb5964a5cb32b180d64d58cec644","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-06 07:51:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/72/302072/1/check/gate-project-config-layout/567f89a/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"51a750ed9b87a8adb1c6f80fa07a74d35648f583","author":{"_account_id":13070,"name":"Eran Gampel","email":"eran@gampel.net","username":"Gampel"},"date":"2016-04-06 08:46:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8cbc5a8e099fbb2e47eb7b51527ef1d49b0dbf1d","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2016-04-06 10:24:56.000000000","message":"Patch Set 1: Code-Review+2\n\nIf the python 3 jobs are failing this will block the project until they are fixed.","accounts_in_message":[],"_revision_number":1},{"id":"2259bd480ee88903306f09ca5845373ff4534997","author":{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},"date":"2016-04-06 11:08:37.000000000","message":"Patch Set 1:\n\nThanks Joshua, we fixed these projects for python3.","accounts_in_message":[],"_revision_number":1},{"id":"26d20277426123118b1170c874fd5c3daa9feedd","author":{"_account_id":17192,"name":"xiangxinyong","email":"xiangxinyong@huawei.com","username":"xianxinyong"},"date":"2016-04-06 11:21:32.000000000","message":"Patch Set 1: Code-Review+1\n\nIt is very good","accounts_in_message":[],"_revision_number":1},{"id":"779b541edf61b640ea6c989a6da6d2f213c0f3de","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2016-04-06 11:54:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fceec5c5203f863cb153836486e37aeb501c5409","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-06 11:54:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"bbadada3d75af147bd7c1edacbf20d4a5dfdcf3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-06 12:06:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/72/302072/1/gate/gate-project-config-layout/2ab87bf/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3bfcd441ab37df198be894f4b29fe5863bf90884","date":"2016-04-06 12:06:41.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4eda300137f1b93f3cce30e26306461aa88c657","revisions":{"a4eda300137f1b93f3cce30e26306461aa88c657":{"kind":"REWORK","_number":1,"created":"2016-04-06 07:44:33.000000000","uploader":{"_account_id":20883,"name":"Yuval Brik","email":"yuval@brik.org.il","username":"yuvalbr"},"ref":"refs/changes/72/302072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/302072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/302072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/302072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/302072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/302072/1"}}},"commit":{"parents":[{"commit":"fb1ac308c24e0e92cbeaa3e2a09eef67c5ea592e","subject":"Merge \"Migrate groups to ubuntu-trusty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb1ac308c24e0e92cbeaa3e2a09eef67c5ea592e"}]}],"author":{"name":"Yuval Brik","email":"yuval@brik.org.il","date":"2016-04-05 13:16:19.000000000","tz":180},"committer":{"name":"Yuval Brik","email":"yuval@brik.org.il","date":"2016-04-06 07:41:47.000000000","tz":180},"subject":"Smaug: add python3 CI jobs","message":"Smaug: add python3 CI jobs\n\nChange-Id: I57c0b06d5d4444dc4c1afbb897612097dac92718\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4eda300137f1b93f3cce30e26306461aa88c657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4eda300137f1b93f3cce30e26306461aa88c657"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
