)]}'
{"id":"openstack%2Fproject-config~325992","triplet_id":"openstack%2Fproject-config~master~I556e15329c793a617ef6d349bce8f420f1fca6d0","project":"openstack/project-config","branch":"master","topic":"v25","hashtags":[],"change_id":"I556e15329c793a617ef6d349bce8f420f1fca6d0","subject":"Revert \"Revert \"Revert \"Revert \"Revert \"Revert \"Start using the Zuul Ansible launcher\"\"\"\"\"\"","status":"MERGED","created":"2016-06-06 15:32:05.000000000","updated":"2016-06-06 15:59:55.000000000","submitted":"2016-06-06 15:59:54.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e9109a9a5aa42c1f7a35812eeedc0f80e03af0e","_number":325992,"virtual_id_number":325992,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2016-06-06 15:59:54.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2016-06-06 15:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-06-06 15:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2016-06-06 15:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-06 15:34:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2016-06-06 15:59:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"45059a3fb3fcbf42e199ffedb34822b226ee8bc8","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2016-06-06 15:32:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b8377b077e6a408eadb5fb894d2ece7fc8204f7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2016-06-06 15:34:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14d57e4d51124197c5ce688d45d54055d730c26f","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2016-06-06 15:39:26.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"41a786abf845c1090dd516123802eb4f059878fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 15:52:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-bindep-fallback-centos-7 http://logs.openstack.org/92/325992/1/check/gate-project-config-bindep-fallback-centos-7/fbdac5d/ : SUCCESS in 1m 20s\n- gate-project-config-bindep-fallback-debian-jessie http://logs.openstack.org/92/325992/1/check/gate-project-config-bindep-fallback-debian-jessie/ad6ccf1/ : SUCCESS in 2m 28s\n- gate-project-config-bindep-fallback-fedora-23 http://logs.openstack.org/92/325992/1/check/gate-project-config-bindep-fallback-fedora-23/f14c01a/ : SUCCESS in 2m 24s\n- gate-project-config-bindep-fallback-ubuntu-precise http://logs.openstack.org/92/325992/1/check/gate-project-config-bindep-fallback-ubuntu-precise/bd65e2c/ : SUCCESS in 2m 41s\n- gate-project-config-bindep-fallback-ubuntu-trusty http://logs.openstack.org/92/325992/1/check/gate-project-config-bindep-fallback-ubuntu-trusty/9e3991c/ : SUCCESS in 1m 16s\n- gate-project-config-bindep-fallback-ubuntu-xenial http://logs.openstack.org/92/325992/1/check/gate-project-config-bindep-fallback-ubuntu-xenial/290cc23/ : SUCCESS in 2m 12s\n- gate-project-config-compare-xml http://logs.openstack.org/92/325992/1/check/gate-project-config-compare-xml/d36e353/ : Jenkins XML output is unchanged. in 11m 52s (non-voting)\n- gate-project-config-gerrit http://logs.openstack.org/92/325992/1/check/gate-project-config-gerrit/1a0c2c1/ : SUCCESS in 41s\n- gate-project-config-grafyaml http://logs.openstack.org/92/325992/1/check/gate-project-config-grafyaml/bd6bda9/ : SUCCESS in 1m 41s\n- gate-project-config-layout http://logs.openstack.org/92/325992/1/check/gate-project-config-layout/4e52bc3/ : SUCCESS in 8m 03s\n- gate-project-config-linters http://logs.openstack.org/92/325992/1/check/gate-project-config-linters/e32c9db/ : SUCCESS in 46s\n- gate-project-config-irc-access http://logs.openstack.org/92/325992/1/check/gate-project-config-irc-access/a26a56f/ : SUCCESS in 2m 57s\n- gate-project-config-jenkins-project http://logs.openstack.org/92/325992/1/check/gate-project-config-jenkins-project/1694561/ : SUCCESS in 50s\n- gate-project-config-nodepool http://logs.openstack.org/92/325992/1/check/gate-project-config-nodepool/30dd911/ : SUCCESS in 41s\n- gate-project-config-projects-yaml http://logs.openstack.org/92/325992/1/check/gate-project-config-projects-yaml/64140f1/ : SUCCESS in 28s\n- gate-infra-docs-index http://logs.openstack.org/92/325992/1/check/gate-infra-docs-index/cd43b24/ : SUCCESS in 37s\n- gate-generate-specs-site http://logs.openstack.org/92/325992/1/check/gate-generate-specs-site/3224587/ : SUCCESS in 44s\n- gate-project-config-dib http://logs.openstack.org/92/325992/1/check/gate-project-config-dib/65035c1/ : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"8d8bd26f9171cd89c2738849a111b9b146730a16","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 15:52:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"bbf4dcb6e12327b0e5615b9cd411a9574a17d835","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 15:59:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-config-bindep-fallback-centos-7 http://logs.openstack.org/92/325992/1/gate/gate-project-config-bindep-fallback-centos-7/1d05b71/ : SUCCESS in 2m 26s\n- gate-project-config-bindep-fallback-debian-jessie http://logs.openstack.org/92/325992/1/gate/gate-project-config-bindep-fallback-debian-jessie/5c6b500/ : SUCCESS in 2m 03s\n- gate-project-config-bindep-fallback-fedora-23 http://logs.openstack.org/92/325992/1/gate/gate-project-config-bindep-fallback-fedora-23/379e420/ : SUCCESS in 1m 46s\n- gate-project-config-bindep-fallback-ubuntu-precise http://logs.openstack.org/92/325992/1/gate/gate-project-config-bindep-fallback-ubuntu-precise/358c852/ : SUCCESS in 1m 53s\n- gate-project-config-bindep-fallback-ubuntu-trusty http://logs.openstack.org/92/325992/1/gate/gate-project-config-bindep-fallback-ubuntu-trusty/e52dffc/ : SUCCESS in 1m 28s\n- gate-project-config-bindep-fallback-ubuntu-xenial http://logs.openstack.org/92/325992/1/gate/gate-project-config-bindep-fallback-ubuntu-xenial/c38b69a/ : SUCCESS in 2m 40s\n- gate-project-config-gerrit http://logs.openstack.org/92/325992/1/gate/gate-project-config-gerrit/cc56028/ : SUCCESS in 1m 08s\n- gate-project-config-grafyaml http://logs.openstack.org/92/325992/1/gate/gate-project-config-grafyaml/019fdee/ : SUCCESS in 1m 02s\n- gate-project-config-layout http://logs.openstack.org/92/325992/1/gate/gate-project-config-layout/8002da0/ : SUCCESS in 5m 06s\n- gate-project-config-linters http://logs.openstack.org/92/325992/1/gate/gate-project-config-linters/81318c1/ : SUCCESS in 55s\n- gate-project-config-irc-access http://logs.openstack.org/92/325992/1/gate/gate-project-config-irc-access/28a9a74/ : SUCCESS in 2m 57s\n- gate-project-config-nodepool http://logs.openstack.org/92/325992/1/gate/gate-project-config-nodepool/f3e8b3c/ : SUCCESS in 1m 02s\n- gate-project-config-jenkins-project http://logs.openstack.org/92/325992/1/gate/gate-project-config-jenkins-project/fa5c54b/ : SUCCESS in 55s\n- gate-project-config-projects-yaml http://logs.openstack.org/92/325992/1/gate/gate-project-config-projects-yaml/0106871/ : SUCCESS in 37s\n- gate-infra-docs-index http://logs.openstack.org/92/325992/1/gate/gate-infra-docs-index/2273fdf/ : SUCCESS in 41s\n- gate-generate-specs-site http://logs.openstack.org/92/325992/1/gate/gate-generate-specs-site/1beb83b/ : SUCCESS in 45s\n- gate-project-config-dib http://logs.openstack.org/92/325992/1/gate/gate-project-config-dib/fff3b69/ : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"39d14edee0cd9390862144a09f8b793ca9e58ccd","date":"2016-06-06 15:59:55.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f5efb9e6ef8ed0909c1ab13878c3ec0f42edbd0","revisions":{"8f5efb9e6ef8ed0909c1ab13878c3ec0f42edbd0":{"kind":"REWORK","_number":1,"created":"2016-06-06 15:32:05.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/92/325992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/325992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/325992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/325992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/325992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/325992/1"}}},"commit":{"parents":[{"commit":"82734616c7a5f09246b1d90328cc0dfe7c220869","subject":"Merge \"Add support for bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82734616c7a5f09246b1d90328cc0dfe7c220869"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2016-06-06 15:31:45.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2016-06-06 15:31:48.000000000","tz":-420},"subject":"Revert \"Revert \"Revert \"Revert \"Revert \"Revert \"Start using the Zuul Ansible launcher\"\"\"\"\"\"","message":"Revert \"Revert \"Revert \"Revert \"Revert \"Revert \"Start using the Zuul Ansible launcher\"\"\"\"\"\"\n\nThis reverts commit d3fc435a9f23281af0ab7faf8d2ade68b177480f.\n\nChange-Id: I556e15329c793a617ef6d349bce8f420f1fca6d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f5efb9e6ef8ed0909c1ab13878c3ec0f42edbd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f5efb9e6ef8ed0909c1ab13878c3ec0f42edbd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
