)]}'
{"id":"openstack%2Fproject-config~475080","triplet_id":"openstack%2Fproject-config~master~If63f602e1e8a4ce3867b99adef5a07c29aa44630","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"If63f602e1e8a4ce3867b99adef5a07c29aa44630","subject":"Add os-service-types project","status":"MERGED","created":"2017-06-16 22:24:11.000000000","updated":"2017-07-07 09:24:57.000000000","submitted":"2017-07-07 09:24:57.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c19289b9b90cf70b8b4684bda82ecef3c979b30f","_number":475080,"virtual_id_number":475080,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2017-07-07 09:24:57.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2017-07-07 02:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-07-07 09:12:36.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-07-07 09:12:36.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-07 02:25:16.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2017-07-07 09:12:36.000000000","updated_by":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"reviewer":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"state":"REVIEWER"},{"updated":"2017-07-07 09:24:57.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e4d4bf6193b7f26944fd0c857fe8bd8081adce58","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2017-06-16 22:24:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1696815542ac33e86ec05fc89b3efa5bd7fcf320","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 22:32:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-gerrit http://logs.openstack.org/80/475080/1/check/gate-project-config-gerrit/8b0573b/ : SUCCESS in 1m 31s\n- gate-project-config-layout http://logs.openstack.org/80/475080/1/check/gate-project-config-layout/5837e3d/ : SUCCESS in 7m 54s\n- gate-project-config-irc-access http://logs.openstack.org/80/475080/1/check/gate-project-config-irc-access/9ea03ac/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"078430f42a237b9580b4530ba7955dadbc8b6de6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-06-18 16:27:29.000000000","message":"Patch Set 1:\n\nWill this be part of any official team or - for now - an independent project just hosted in OpenStack?","accounts_in_message":[],"_revision_number":1},{"id":"87a9ac0254e2bb51432a0ab31a398a977b4f2859","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-07-02 17:12:12.000000000","message":"Topic set to new-project","accounts_in_message":[],"_revision_number":1},{"id":"f47c8dde6a6a85b9170d1bf98074bc84b0cfd2c5","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2017-07-05 20:16:57.000000000","message":"Patch Set 1:\n\nIt\u0027s an independent project just hosted in openstack for now (we can sort out governance ownership later I think)","accounts_in_message":[],"_revision_number":1},{"id":"bf35985b133c3dfdcf9d291809adc8b2dc0a7118","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-07-05 20:18:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6295b8bd697d69333f8150add5b971b06b362488","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2017-07-05 20:22:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74f892901c756a51ccec0191fe43851918b3c93e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-07-05 20:24:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nchange is fine - but since you publish to docs.o.o, you need a governance change","accounts_in_message":[],"_revision_number":2},{"id":"3ebb57a1411a1dc1abd5b88ccafa17f9209520c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-05 20:30:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-project-config-gerrit http://logs.openstack.org/80/475080/2/check/gate-project-config-gerrit/84fef8d/ : SUCCESS in 1m 01s\n- gate-project-config-layout http://logs.openstack.org/80/475080/2/check/gate-project-config-layout/4f534b3/ : FAILURE in 1m 12s\n- gate-project-config-irc-access http://logs.openstack.org/80/475080/2/check/gate-project-config-irc-access/ed3946e/ : SUCCESS in 4m 29s\n- gate-project-config-jenkins-project http://logs.openstack.org/80/475080/2/check/gate-project-config-jenkins-project/de27d49/ : FAILURE in 1m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cffa778044d12e5e7d58d9ac525e1ab20630dc01","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2017-07-05 21:32:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1f5c2a097a5988462ae2cbb00775b6b35854033","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-05 21:44:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-gerrit http://logs.openstack.org/80/475080/3/check/gate-project-config-gerrit/f4f0b1a/ : SUCCESS in 2m 25s\n- gate-project-config-layout http://logs.openstack.org/80/475080/3/check/gate-project-config-layout/94e2257/ : SUCCESS in 11m 48s\n- gate-project-config-irc-access http://logs.openstack.org/80/475080/3/check/gate-project-config-irc-access/353ff6e/ : SUCCESS in 6m 05s\n- gate-project-config-jenkins-project http://logs.openstack.org/80/475080/3/check/gate-project-config-jenkins-project/35b5095/ : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":3},{"id":"d6ad3128cdbd468e6638deaf9abe310c9b295e92","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-07-06 04:56:37.000000000","message":"Patch Set 3:\n\nhttps://pypi.python.org/pypi/os-service-types does not exist, please create.\n\nOnce that\u0027s set up, I\u0027ll +2.","accounts_in_message":[],"_revision_number":3},{"id":"9620a648b66af91405558f81943bc7c27865ad59","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2017-07-06 21:42:55.000000000","message":"Patch Set 3:\n\nDone! I also added openstackci as an owner.","accounts_in_message":[],"_revision_number":3},{"id":"e9377d04366eac6dfbb043d3bfecbe6b2a2953c8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-07-07 02:25:16.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, PTL has submitted","accounts_in_message":[],"_revision_number":3},{"id":"b529a5be45fa3873f38556d13ea744c227824cd4","author":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"date":"2017-07-07 09:12:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb4c6e92892cb74762108a65f3c079003aeebc6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-07 09:15:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"4688f074849dfa6807e8caa83134486911c81dea","date":"2017-07-07 09:24:57.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"c19289b9b90cf70b8b4684bda82ecef3c979b30f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-07 09:24:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-config-gerrit http://logs.openstack.org/80/475080/3/gate/gate-project-config-gerrit/04aab4e/ : SUCCESS in 1m 59s\n- gate-project-config-layout http://logs.openstack.org/80/475080/3/gate/gate-project-config-layout/a3666de/ : SUCCESS in 6m 08s\n- gate-project-config-irc-access http://logs.openstack.org/80/475080/3/gate/gate-project-config-irc-access/8d0ed9f/ : SUCCESS in 4m 26s\n- gate-project-config-jenkins-project http://logs.openstack.org/80/475080/3/gate/gate-project-config-jenkins-project/1a1283c/ : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b7d966a1f4bf7d08613f89f0e92f62ec26ff6a06","revisions":{"8001e373628b086ff04b6c44df246d3e31bfdb18":{"kind":"REWORK","_number":1,"created":"2017-06-16 22:24:11.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/80/475080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/475080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/475080/1"}}},"commit":{"parents":[{"commit":"ad47cac9a5d18b65401fa39c3274bd11a00a8657","subject":"Merge \"Stop running python2 jobs for Zuul v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad47cac9a5d18b65401fa39c3274bd11a00a8657"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2017-06-16 22:22:42.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2017-06-16 22:23:57.000000000","tz":-300},"subject":"Add os-service-types project","message":"Add os-service-types project\n\nThe API-WG recently defined the publication of data about the official\nservice types and their historical aliases. os-service-types is a\nlibrary to consume and encapsulate that data properly for anyone using\npython.\n\nChange-Id: If63f602e1e8a4ce3867b99adef5a07c29aa44630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8001e373628b086ff04b6c44df246d3e31bfdb18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8001e373628b086ff04b6c44df246d3e31bfdb18"}]},"branch":"refs/heads/master"},"63fea0edfaf452f3d7f2fe346323a8a3c0bc587c":{"kind":"REWORK","_number":2,"created":"2017-07-05 20:22:33.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/80/475080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/475080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/475080/2"}}},"commit":{"parents":[{"commit":"d94fd80f1d2ef87d64e8483d5f56107e1d113ab0","subject":"Merge \"Remove grenade-nova-net template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d94fd80f1d2ef87d64e8483d5f56107e1d113ab0"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2017-06-16 22:22:42.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2017-07-05 20:21:32.000000000","tz":-300},"subject":"Add os-service-types project","message":"Add os-service-types project\n\nThe API-WG recently defined the publication of data about the official\nservice types and their historical aliases. os-service-types is a\nlibrary to consume and encapsulate that data properly for anyone using\npython.\n\nAddition to official governance will be handled later. For now it\u0027s just\na project in the infrastructure.\n\nChange-Id: If63f602e1e8a4ce3867b99adef5a07c29aa44630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/63fea0edfaf452f3d7f2fe346323a8a3c0bc587c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/63fea0edfaf452f3d7f2fe346323a8a3c0bc587c"}]},"branch":"refs/heads/master"},"b7d966a1f4bf7d08613f89f0e92f62ec26ff6a06":{"kind":"REWORK","_number":3,"created":"2017-07-05 21:32:29.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/80/475080/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/475080/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/475080/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/475080/3"}}},"commit":{"parents":[{"commit":"d94fd80f1d2ef87d64e8483d5f56107e1d113ab0","subject":"Merge \"Remove grenade-nova-net template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d94fd80f1d2ef87d64e8483d5f56107e1d113ab0"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2017-06-16 22:22:42.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2017-07-05 21:32:23.000000000","tz":-300},"subject":"Add os-service-types project","message":"Add os-service-types project\n\nThe API-WG recently defined the publication of data about the official\nservice types and their historical aliases. os-service-types is a\nlibrary to consume and encapsulate that data properly for anyone using\npython.\n\nNeeded-By: I7621760d16c282c8671580e4f7be17ee1699e511\nChange-Id: If63f602e1e8a4ce3867b99adef5a07c29aa44630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7d966a1f4bf7d08613f89f0e92f62ec26ff6a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7d966a1f4bf7d08613f89f0e92f62ec26ff6a06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
