)]}'
{"id":"openstack%2Fproject-config~493758","triplet_id":"openstack%2Fproject-config~master~Ifebe04d9fd585b78ab14cd7ac4b8b69549fa4356","project":"openstack/project-config","branch":"master","topic":"py35_5","hashtags":[],"change_id":"Ifebe04d9fd585b78ab14cd7ac4b8b69549fa4356","subject":"Add Vitrage python 35 jobs as non-voting","status":"MERGED","created":"2017-08-15 06:46:22.000000000","updated":"2017-08-16 08:18:32.000000000","submitted":"2017-08-16 08:18:32.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e9f540aee2307e98aaba429ecd6ecd4f95d07e3","_number":493758,"virtual_id_number":493758,"owner":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-08-16 08:18:32.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-08-15 10:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2017-08-16 08:06:43.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"value":1,"date":"2017-08-16 07:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2017-08-16 08:06:43.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-15 10:32:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2017-08-16 07:42:53.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2017-08-16 08:06:43.000000000","updated_by":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"reviewer":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"state":"REVIEWER"},{"updated":"2017-08-16 08:18:32.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ebb3ba10fced3b44477d0cd18528636cb6e07ae2","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-08-15 06:46:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01754cb2522790d039cc2f4facee83adacfc9746","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2017-08-15 06:50:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1067b309894095048fbbda049db5a74c6c654c1b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-15 06:54:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-project-config-layout http://logs.openstack.org/58/493758/1/check/gate-project-config-layout/9052e31/ : FAILURE in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f82473b7976a79828ac472e2e2dcfe1a55866676","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-08-15 07:31:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b19be227342842a1a8907b25dc5d4e7ed25fd7b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-15 07:39:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-project-config-layout http://logs.openstack.org/58/493758/2/check/gate-project-config-layout/8fdc7b1/ : FAILURE in 7m 57s","accounts_in_message":[],"_revision_number":2},{"id":"7603da7fcaef637bdbb9d88d0af3e5bef47bb8d4","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-08-15 07:54:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"44e595b649be904e3716b99c64258fb7fdddeba6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-15 08:02:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/58/493758/3/check/gate-project-config-layout/1058e41/ : SUCCESS in 7m 23s\n- gate-project-config-jenkins-project http://logs.openstack.org/58/493758/3/check/gate-project-config-jenkins-project/7fc82e5/ : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":3},{"id":"2bc1c5db9ec927bbff559760e4b8c7e1591296b5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-08-15 08:18:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dcfeae61294690995869425a8842f09e54b757bc","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-08-15 08:23:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5bb38aee8d16e3d7f5cbfcb55c871929a2f2ae9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-15 08:32:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/58/493758/4/check/gate-project-config-layout/fcfa50a/ : SUCCESS in 8m 22s\n- gate-project-config-jenkins-project http://logs.openstack.org/58/493758/4/check/gate-project-config-jenkins-project/af1ea9d/ : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f9bfe7c2c7b400c84d0eca6ad31edff37a760537","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-08-15 10:32:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f71b2010f94e52b606693d341f81cf72aea4c9f6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2017-08-16 07:42:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a413b455fac809cff7aaee39ade8af0dfd56b468","author":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"date":"2017-08-16 08:06:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4441ae3463f6e7906084f9564059bafd1436a898","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-16 08:06:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"4b63912a444229787d6a53ad9b357b1725bcffbc","date":"2017-08-16 08:18:32.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":4},{"id":"373567b3665c4632169b42b14f3b4890996a3b1d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-16 08:18:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/58/493758/4/gate/gate-project-config-layout/de6372b/ : SUCCESS in 11m 22s\n- gate-project-config-jenkins-project http://logs.openstack.org/58/493758/4/gate/gate-project-config-jenkins-project/4209408/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"520220de0159164eb3cd955013c499e02452b61b","revisions":{"60e0ad92e4476c240b8a0d1654f398f73e6cfd02":{"kind":"REWORK","_number":1,"created":"2017-08-15 06:46:22.000000000","uploader":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"ref":"refs/changes/58/493758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/58/493758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/58/493758/1"}}},"commit":{"parents":[{"commit":"b1e969204c70dc002670d9afc8ade336fb13d985","subject":"Merge \"Update NODEPOOL_RUGYGEMS_PROXY / add bundle client support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1e969204c70dc002670d9afc8ade336fb13d985"}]}],"author":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 06:44:26.000000000","tz":0},"committer":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 06:46:19.000000000","tz":0},"subject":"Add Vitrage python 35 jobs as non-voting","message":"Add Vitrage python 35 jobs as non-voting\n\nChange-Id: Ifebe04d9fd585b78ab14cd7ac4b8b69549fa4356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60e0ad92e4476c240b8a0d1654f398f73e6cfd02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60e0ad92e4476c240b8a0d1654f398f73e6cfd02"}]},"branch":"refs/heads/master"},"88f89f9c21c0631f794a4fa7359b3d7629aa8dc7":{"kind":"REWORK","_number":2,"created":"2017-08-15 07:31:02.000000000","uploader":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"ref":"refs/changes/58/493758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/58/493758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/58/493758/2"}}},"commit":{"parents":[{"commit":"b1e969204c70dc002670d9afc8ade336fb13d985","subject":"Merge \"Update NODEPOOL_RUGYGEMS_PROXY / add bundle client support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1e969204c70dc002670d9afc8ade336fb13d985"}]}],"author":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 06:44:26.000000000","tz":0},"committer":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 07:30:04.000000000","tz":0},"subject":"Add Vitrage python 35 jobs as non-voting","message":"Add Vitrage python 35 jobs as non-voting\n\nChange-Id: Ifebe04d9fd585b78ab14cd7ac4b8b69549fa4356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88f89f9c21c0631f794a4fa7359b3d7629aa8dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88f89f9c21c0631f794a4fa7359b3d7629aa8dc7"}]},"branch":"refs/heads/master"},"0f993a9f3fe11db1c227cc78b144b77910e1ffc1":{"kind":"REWORK","_number":3,"created":"2017-08-15 07:54:56.000000000","uploader":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"ref":"refs/changes/58/493758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/58/493758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/58/493758/3"}}},"commit":{"parents":[{"commit":"b1e969204c70dc002670d9afc8ade336fb13d985","subject":"Merge \"Update NODEPOOL_RUGYGEMS_PROXY / add bundle client support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1e969204c70dc002670d9afc8ade336fb13d985"}]}],"author":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 06:44:26.000000000","tz":0},"committer":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 07:54:32.000000000","tz":0},"subject":"Add Vitrage python 35 jobs as non-voting","message":"Add Vitrage python 35 jobs as non-voting\n\nChange-Id: Ifebe04d9fd585b78ab14cd7ac4b8b69549fa4356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f993a9f3fe11db1c227cc78b144b77910e1ffc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f993a9f3fe11db1c227cc78b144b77910e1ffc1"}]},"branch":"refs/heads/master"},"520220de0159164eb3cd955013c499e02452b61b":{"kind":"REWORK","_number":4,"created":"2017-08-15 08:23:42.000000000","uploader":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"ref":"refs/changes/58/493758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/58/493758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/493758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/58/493758/4"}}},"commit":{"parents":[{"commit":"b1e969204c70dc002670d9afc8ade336fb13d985","subject":"Merge \"Update NODEPOOL_RUGYGEMS_PROXY / add bundle client support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1e969204c70dc002670d9afc8ade336fb13d985"}]}],"author":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 06:44:26.000000000","tz":0},"committer":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-08-15 08:23:16.000000000","tz":0},"subject":"Add Vitrage python 35 jobs as non-voting","message":"Add Vitrage python 35 jobs as non-voting\n\nChange-Id: Ifebe04d9fd585b78ab14cd7ac4b8b69549fa4356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/520220de0159164eb3cd955013c499e02452b61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/520220de0159164eb3cd955013c499e02452b61b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
