)]}'
{"id":"openstack%2Fproject-config~506332","triplet_id":"openstack%2Fproject-config~master~I4f3265c16320613d4ba74a02df1361c5d9cf2fb1","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I4f3265c16320613d4ba74a02df1361c5d9cf2fb1","subject":"Update logfile for unbound on centos/fedora","status":"MERGED","created":"2017-09-21 19:23:47.000000000","updated":"2017-09-22 08:27:11.000000000","submitted":"2017-09-22 08:27:11.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"506332-1506068831948-2a1f4ec0","meta_rev_id":"ae481144fe4b7ae5921975c180b300be23d73bf7","_number":506332,"virtual_id_number":506332,"owner":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2017-09-22 08:27:11.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2017-09-21 19:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-09-22 08:18:30.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-09-22 08:18:30.000000000","_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-21 19:27:11.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2017-09-22 08:18:30.000000000","updated_by":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"reviewer":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"state":"REVIEWER"},{"updated":"2017-09-22 08:27:11.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"fdd6a592e509ba92485626a12f773d5122c9dc46","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2017-09-21 19:23:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b77fe3ca997d2b453f2dbe277730928028c6996","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-09-21 19:27:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cf58b5bde98aa65d2baa380d5678bdd4a067528","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 19:33:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/32/506332/1/check/gate-project-config-layout/5d3284e/ : SUCCESS in 8m 13s\n- gate-project-config-dib http://logs.openstack.org/32/506332/1/check/gate-project-config-dib/98f5147/ : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bf97ddbaf71266c824f0cf77ca30e6f4fa7ac7cf","author":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"date":"2017-09-22 08:18:30.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84252855e1fab25692c4066226a09d9dc1eedabd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 08:18:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"0e2c8a6f1fb74ddfaae4b353b7d9d9d0d7b94549","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 08:27:11.000000000","message":"Change has been successfully merged by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"ae481144fe4b7ae5921975c180b300be23d73bf7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 08:27:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-config-layout http://logs.openstack.org/32/506332/1/gate/gate-project-config-layout/6ed4c9b/ : SUCCESS in 8m 14s\n- gate-project-config-dib http://logs.openstack.org/32/506332/1/gate/gate-project-config-dib/14970ff/ : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb25f0b71635e0061b9b65a51d1462438b6a862e","revisions":{"bb25f0b71635e0061b9b65a51d1462438b6a862e":{"kind":"REWORK","_number":1,"created":"2017-09-21 19:23:47.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/32/506332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/506332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/506332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/506332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/506332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/506332/1"}}},"commit":{"parents":[{"commit":"d2fee232d2a486a0e062298fb92628697c045201","subject":"Merge \"Increase timeouts for Manila DSVM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d2fee232d2a486a0e062298fb92628697c045201"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2017-09-21 19:00:10.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2017-09-21 19:00:10.000000000","tz":-240},"subject":"Update logfile for unbound on centos/fedora","message":"Update logfile for unbound on centos/fedora\n\nBecause of missing selinux rules, unbound does not properly write\nlogging to /var/log/unbound.log. We can move the file into\n/var/lib/unbound/unbound.log and selinux should be happy.\n\nChange-Id: I4f3265c16320613d4ba74a02df1361c5d9cf2fb1\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb25f0b71635e0061b9b65a51d1462438b6a862e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb25f0b71635e0061b9b65a51d1462438b6a862e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
