)]}'
{"id":"openstack%2Fproject-config~513506","triplet_id":"openstack%2Fproject-config~master~I912bf6ce941a92bdfc37f50f267fda892e2407de","project":"openstack/project-config","branch":"master","topic":"zuulv3-jobs","hashtags":[],"change_id":"I912bf6ce941a92bdfc37f50f267fda892e2407de","subject":"Remove fetch-zuul-cloner from base job","status":"ABANDONED","created":"2017-10-19 20:53:36.000000000","updated":"2019-02-06 21:38:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"775ba36c427990aea578c772c1d18a37cb22a1e2","_number":513506,"virtual_id_number":513506,"owner":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"date":"2018-03-07 02:44:56.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"date":"2018-03-07 06:10:51.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":1,"date":"2018-03-07 02:49:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":-2,"date":"2018-03-07 02:44:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-23 22:27:07.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2017-10-31 15:28:14.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2017-11-29 16:07:23.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2018-03-07 02:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-07 06:10:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"4f54c2ee8a53806e3b306282a64e9fc9e5e83858","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2017-10-19 20:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4abd35293df50d3e8dee6637e8a7f7cb25454de7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 21:01:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-linters http://logs.openstack.org/06/513506/1/infra-check/openstack-tox-linters/24338c8/ : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e334534b9a6dfa318bd6886c9d5b9a11fa0b448f","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2017-10-19 21:02:09.000000000","message":"Patch Set 1:\n\nbase-test (no fetch-zuul-cloner )\n  http://logs.openstack.org/02/513502/1/check/tox-linters/14d6afa/job-output.txt.gz\n\nlegacy-base (fetch-zuul-cloner)\n  http://logs.openstack.org/54/513054/9/check/windmill-ubuntu-xenial/bdad98b/job-output.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"b72a64fb3d7eb406c19b405b03c5e3bf992d92ba","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2017-10-20 11:33:07.000000000","message":"Topic set to zuulv3-jobs","accounts_in_message":[],"_revision_number":1},{"id":"4aedbfcf9f091cf355ab65af0d437b7e117ef927","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2017-10-23 22:27:07.000000000","message":"Patch Set 1: Code-Review-1\n\nLet\u0027s merge this only after we\u0027ve stopped including zuul-cloner on images.","accounts_in_message":[],"_revision_number":1},{"id":"45fc2e15f0af6c0463da9fe10eabc8051acb63e8","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2017-10-31 15:28:14.000000000","message":"Patch Set 1: Code-Review-1\n\nWe need to be careful with this.\nThe fact that a job is *not* legacy doesn\u0027t mean it doesn\u0027t use zuul-cloner.\n\nNew, zuul v3 native jobs can end up using zuul-cloner somewhere down the line, for example inside projects\u0027 scripts in their repositories [1].\n\n[1]: http://codesearch.openstack.org/?q\u003d%5C%24ZUUL_CLONER\u0026i\u003dnope\u0026files\u003dtools%2Ftox_install.sh\u0026repos\u003d","accounts_in_message":[],"_revision_number":1},{"id":"b36d5eb8eb539fa19c866790abcf2b4af62b7091","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2017-10-31 15:45:12.000000000","message":"Patch Set 1:\n\njobs that need zuul-cloner, should parent to base-legacy, IMO that is our long term path forward. So, if zuulv3 native jobs are still using zuul-cloner, this is a bug we need to fix.\n\nHow we fix it is likely up for debate, either projects can create a variant and setup fetch-zuul-cloner role in pre-run or create a compatible legacy- job which just patents to base-legacy.","accounts_in_message":[],"_revision_number":1},{"id":"913249ea83787140309a68843599d543e34d1ab0","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2017-11-21 19:32:47.000000000","message":"Patch Set 1:\n\nAs discussed in meeting:\n\n19:31 \u003c jeblair\u003e #info 1) remove zuul-env from images now.   if that doesn\u0027t blow up:\n19:31 \u003c jeblair\u003e #info 2) remove zuul-cloner shim from base immediately afterwords.\n19:31 \u003c jeblair\u003e #info 3) if that blows up, fix those jobs, or temporarily revert 506 to add zuul-cloner shim back.  repeat as necessary.\n19:31 \u003c jeblair\u003e #info 4) in the distant future, remove legacy-base (which will continue to install the shim as long as it exists)","accounts_in_message":[],"_revision_number":1},{"id":"c8a18143f452508e870fe5776065fa0d65e3f230","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2017-11-29 16:07:23.000000000","message":"Patch Set 1: Code-Review-2\n\nThis cannot land until https://etherpad.openstack.org/p/zuulv3-remove-zuul-cloner has been taken care of.","accounts_in_message":[],"_revision_number":1},{"id":"5016ba37d895118768cf49a6904031509de5d152","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2018-03-07 02:44:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63cc0100f9632cd62362611f9df0648e5a0ae062","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2018-03-07 02:44:56.000000000","message":"Patch Set 2:\n\nEverything outlined my monty should be done","accounts_in_message":[],"_revision_number":2},{"id":"c850681ecc90ef2ce5394ec3ac620889cfb5fc9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 02:49:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/06/513506/2/check/openstack-zuul-jobs-linters/578d41a/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"2599283a15e19b28099bbfbbfa688ae707085482","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-03-07 06:10:51.000000000","message":"Patch Set 2:\n\nNo, it\u0027s not done - see lines 16 and following. This is not complete.\n\nMonty, what are next steps here?","accounts_in_message":[],"_revision_number":2},{"id":"a29e2cf2901f11a89fb9d964ba3e4d3904874ab6","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2018-09-08 00:52:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"21170f29d996f73bcf8950f1e5b144b6f1960694","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2018-12-17 19:40:51.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"696ebdb0cb0055ad03eada241abbed06acad4a40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-17 19:49:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/06/513506/2/check/openstack-zuul-jobs-linters/68e0c1f/ : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"775ba36c427990aea578c772c1d18a37cb22a1e2","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-02-06 21:38:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e82b5f13edcd1e787a9392286c1402dada675e7","revisions":{"fb32bcbab4f372b3d5ffe1f153e887f22ee7d240":{"kind":"REWORK","_number":1,"created":"2017-10-19 20:53:36.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/06/513506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/513506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/513506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/513506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/513506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/513506/1"}}},"commit":{"parents":[{"commit":"69375f013f9833ee30726b13ccbf218dafc8fea0","subject":"Merge \"Remove double python in translation script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69375f013f9833ee30726b13ccbf218dafc8fea0"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2017-10-19 20:52:31.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2017-10-19 20:53:20.000000000","tz":-240},"subject":"Remove fetch-zuul-cloner from base job","message":"Remove fetch-zuul-cloner from base job\n\nThis has been moved into legacy-base, as it represents zuulv2\nfunctionality.  By moving it into legacy-base, we discourage native\nzuulv3 jobs from using it, as it is not needed.\n\nThis promotes: If32a373cdb8a4899ccaa00d3c7df32fe76574f80\n\nChange-Id: I912bf6ce941a92bdfc37f50f267fda892e2407de\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb32bcbab4f372b3d5ffe1f153e887f22ee7d240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb32bcbab4f372b3d5ffe1f153e887f22ee7d240"}]},"branch":"refs/heads/master"},"3e82b5f13edcd1e787a9392286c1402dada675e7":{"kind":"REWORK","_number":2,"created":"2018-03-07 02:44:28.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/06/513506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/513506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/513506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/513506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/513506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/513506/2"}}},"commit":{"parents":[{"commit":"0bf4f09e1184546d2d0ddbbfe41184eb37350f4b","subject":"Merge \"Remove k8s-openstack-provider jobs for repo refresh.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0bf4f09e1184546d2d0ddbbfe41184eb37350f4b"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2017-10-19 20:52:31.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2018-03-07 02:44:12.000000000","tz":-300},"subject":"Remove fetch-zuul-cloner from base job","message":"Remove fetch-zuul-cloner from base job\n\nThis has been moved into legacy-base, as it represents zuulv2\nfunctionality.  By moving it into legacy-base, we discourage native\nzuulv3 jobs from using it, as it is not needed.\n\nThis promotes: If32a373cdb8a4899ccaa00d3c7df32fe76574f80\n\nChange-Id: I912bf6ce941a92bdfc37f50f267fda892e2407de\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e82b5f13edcd1e787a9392286c1402dada675e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e82b5f13edcd1e787a9392286c1402dada675e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
