)]}'
{"id":"openstack%2Fproject-config~533608","triplet_id":"openstack%2Fproject-config~master~I56d44f8dff41dbf3b2ff2382fa39b364f55f9a44","project":"openstack/project-config","branch":"master","topic":"nova-tox-functional","hashtags":[],"change_id":"I56d44f8dff41dbf3b2ff2382fa39b364f55f9a44","subject":"Moving nova functional test def to in tree","status":"MERGED","created":"2018-01-15 10:55:54.000000000","updated":"2018-01-31 15:20:53.000000000","submitted":"2018-01-31 15:20:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"533608-1517412053209-49b85806","meta_rev_id":"79bf673eb8d14be97aa540b0cf17552f904b0bb0","_number":533608,"virtual_id_number":533608,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2018-01-31 15:03:37.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2018-01-31 15:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":2,"date":"2018-01-31 15:06:28.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":2,"date":"2018-01-29 20:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-01-31 15:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":1,"date":"2018-01-31 15:06:28.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-16 08:55:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2018-01-29 20:37:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-01-31 15:06:08.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-01-31 15:06:28.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2018-01-31 15:20:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0bfa7eff5d5f86f6f7fbf02b4e3563dd025101a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-15 10:55:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2af7620705a5a65c8102d8f59411c2d751d03973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-15 10:56:34.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5a22945e4448f583061caed464ffc4d5e8ad397b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-01-15 11:49:08.000000000","message":"Patch Set 1:\n\nchange LGTM - but rebase it, please.","accounts_in_message":[],"_revision_number":1},{"id":"7182a91cc9cecf901b2217bf0428c4c21903e104","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-15 12:07:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bfa39466316627fe4c1e109e0249e1055e85a216","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-15 12:07:39.000000000","message":"Patch Set 1:\n\n\u003e change LGTM - but rebase it, please.\nDone","accounts_in_message":[],"_revision_number":1},{"id":"2c0cfff964f399a7f1346462bac38e11dc21dd0a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-01-15 12:13:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c0ac158487233888c01bce0e24ada369608123cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-15 12:24:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/08/533608/2/check/openstack-zuul-jobs-linters/70a5569/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3f5ef8ca2cf8d8a5d7e3619501ea4255b0df1b9f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-01-16 08:55:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fbe7e04261a73da944c0d0b8ccfa22c6d3ce91d6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-16 09:08:12.000000000","message":"Patch Set 2: Workflow-1\n\nWe need to wait this removal unit the backports of Ia684786d1622da7af31aa4479fc883a7c65848ff are merged to the stable branches in nova.","accounts_in_message":[],"_revision_number":2},{"id":"1c29d9be225135f55da0eb9f47e69e70cc840776","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-23 19:14:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ceed40143d9fe008f2a685f43131a488f3f66f0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-23 19:20:42.000000000","message":"Patch Set 2: Code-Review+1\n\nStable branch backports have been posted.","accounts_in_message":[],"_revision_number":2},{"id":"64c81586c432d36955db0a5e6caa63613ba5b583","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-28 00:57:12.000000000","message":"Patch Set 2:\n\nChanges in nova are all approved, let\u0027s do this.","accounts_in_message":[],"_revision_number":2},{"id":"20a6b74213b556fd0c3e8960f557f3ac7b8b61d7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-01-28 10:27:56.000000000","message":"Patch Set 2:\n\nwe run in timeouts with this job and should remove it. Let me rebase to remove the WIP.","accounts_in_message":[],"_revision_number":2},{"id":"aba7f1ee20118602f526aac28bfb0603c9d55c54","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-01-28 10:27:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"342aaebe9503bcc282ecf0c8c9013aaa741207cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-28 10:36:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/08/533608/3/check/openstack-zuul-jobs-linters/ade91f8/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"247af6875a899268e49dea2eea3dd8c6d879e21e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-28 14:58:59.000000000","message":"Patch Set 3:\n\nHaving a hard time getting the nova changes to merge on stable because the cellsv1 job continually keeps failing on stable for some reason, see bug 1745838. I might have to push a patch to no longer run the cellsv1 job on stable.","accounts_in_message":[],"_revision_number":3},{"id":"1d29ad4e131b8c2dadc31d24db0d17dafdeeeed2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-28 15:59:02.000000000","message":"Patch Set 3:\n\nWe might have to do this:\n\nhttps://review.openstack.org/#/c/538619/","accounts_in_message":[],"_revision_number":3},{"id":"3095dafed1779fe9d7d92c45ac7a57dcbafd5346","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-29 20:25:24.000000000","message":"Patch Set 3:\n\nThe dependencies in the stable branches are failing to merge because they are hitting timeouts from the nova-tox-functional jobs they are trying to add, and we can\u0027t land the timeout patch on master until this mess is merged. I could backport the timeout change to stable and squash those changes, but that gets messy too. So I think I\u0027ll change this to not remove but restrict these jobs to only run on stable for nova, and then I\u0027ll push the removal patch on top and that can depend on the nova stable branch changes.","accounts_in_message":[],"_revision_number":3},{"id":"b1d05b45af7a14340c9328fda8501f1b461b2e50","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-29 20:34:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"954cf3a65642b2b403dfa0334ed89a0a8feead31","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-01-29 20:37:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ac842d721649c5933d9c95b4e86c4b258550c1f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-29 22:22:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/08/533608/4/check/openstack-zuul-jobs-linters/4a3fb08/ : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":4},{"id":"444b9fde8b8aa190ab4406fe90075bc59c075ff4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-31 15:03:37.000000000","message":"Patch Set 4:\n\nThe ocata backport has been merged as well in https://review.openstack.org/#/c/536936/ so this is now good to go in.","accounts_in_message":[],"_revision_number":4},{"id":"4872e7bd446c5faec719a33c2ff9774b065885b7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-31 15:06:08.000000000","message":"Patch Set 4: Code-Review+1\n\ndeps are all merged","accounts_in_message":[],"_revision_number":4},{"id":"cb8e1e617be307dbc6dc54fdad3dc957f57734ab","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2018-01-31 15:06:28.000000000","message":"Patch Set 4: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5fe6ff1ab89fa046ae85bd07120e5dacbc63e581","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 15:07:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4441bd409e92de1e156a849d15d1ed2f16859145","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 15:20:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/08/533608/4/gate/openstack-zuul-jobs-linters/23068c1/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"a30af8a85e8987c3bedc876239f2b3c4980bc83a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 15:20:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8ccbe79648af49a135da23af9c543f0876dd2e40","revisions":{"3f2cb0129fe8918f8668de0fe7be740bc733e050":{"kind":"REWORK","_number":1,"created":"2018-01-15 10:55:54.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/08/533608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/533608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/533608/1"}}},"commit":{"parents":[{"commit":"78a731802b456a0f2684c083089b9638ab94c061","subject":"Merge \"Co-gate tempest-plugins with main repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/78a731802b456a0f2684c083089b9638ab94c061"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-15 10:46:10.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-15 10:55:49.000000000","tz":60},"subject":"Moving nova functional test def to in tree","message":"Moving nova functional test def to in tree\n\nA new functional job definition is added in the nova tree in\nIa684786d1622da7af31aa4479fc883a7c65848ff to make sure that the\nirrelevant-files is properly overridden. Therefore\nopenstack-tox-functional and functional-py35 is removed from the nova\npart of the project-config.\n\nDepends-On: Ia684786d1622da7af31aa4479fc883a7c65848ff\nChange-Id: I56d44f8dff41dbf3b2ff2382fa39b364f55f9a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f2cb0129fe8918f8668de0fe7be740bc733e050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f2cb0129fe8918f8668de0fe7be740bc733e050"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"78a731802b456a0f2684c083089b9638ab94c061","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"caa6866b1f13627e912de7945b3aa4035fcbe5d7":{"kind":"REWORK","_number":2,"created":"2018-01-15 12:07:20.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/08/533608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/533608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/533608/2"}}},"commit":{"parents":[{"commit":"d08d391ba048c7476860120cca17b5d10d4ba4f6","subject":"Rename neutron jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d08d391ba048c7476860120cca17b5d10d4ba4f6"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-15 10:46:10.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-15 12:07:06.000000000","tz":60},"subject":"Moving nova functional test def to in tree","message":"Moving nova functional test def to in tree\n\nA new functional job definition is added in the nova tree in\nIa684786d1622da7af31aa4479fc883a7c65848ff to make sure that the\nirrelevant-files is properly overridden. Therefore\nopenstack-tox-functional and functional-py35 is removed from the nova\npart of the project-config.\n\nDepends-On: Ia684786d1622da7af31aa4479fc883a7c65848ff\nChange-Id: I56d44f8dff41dbf3b2ff2382fa39b364f55f9a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/caa6866b1f13627e912de7945b3aa4035fcbe5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/caa6866b1f13627e912de7945b3aa4035fcbe5d7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d08d391ba048c7476860120cca17b5d10d4ba4f6","is_merged_in_target_branch":true,"change_id":"I3d0f9369a87a9e61c611a2323f68e68051d5ed7d","change_number":533348,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0de284f147a388bfbf10e74f3be3d935b7faeda6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-01-28 10:27:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/08/533608/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/533608/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/533608/3"}}},"commit":{"parents":[{"commit":"b7c0a786ac1178c10d232408c1a0a287c73e83d3","subject":"Temporarily remove ansible/ansible from pipelines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7c0a786ac1178c10d232408c1a0a287c73e83d3"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-15 10:46:10.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2018-01-28 10:27:59.000000000","tz":0},"subject":"Moving nova functional test def to in tree","message":"Moving nova functional test def to in tree\n\nA new functional job definition is added in the nova tree in\nIa684786d1622da7af31aa4479fc883a7c65848ff to make sure that the\nirrelevant-files is properly overridden. Therefore\nopenstack-tox-functional and functional-py35 is removed from the nova\npart of the project-config.\n\nDepends-On: Ia684786d1622da7af31aa4479fc883a7c65848ff\nChange-Id: I56d44f8dff41dbf3b2ff2382fa39b364f55f9a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0de284f147a388bfbf10e74f3be3d935b7faeda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0de284f147a388bfbf10e74f3be3d935b7faeda6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b7c0a786ac1178c10d232408c1a0a287c73e83d3","is_merged_in_target_branch":true,"change_id":"Ic0654aeb332a89235baab05740c0843ee7de2a47","change_number":538447,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8ccbe79648af49a135da23af9c543f0876dd2e40":{"kind":"REWORK","_number":4,"created":"2018-01-29 20:34:47.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/08/533608/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/533608/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/533608/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/533608/4"}}},"commit":{"parents":[{"commit":"d57cd43bf95f055320f365897511b285f80052c4","subject":"Only run openstack-tox-functional on nova stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d57cd43bf95f055320f365897511b285f80052c4"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-15 10:46:10.000000000","tz":60},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-01-29 20:34:32.000000000","tz":-300},"subject":"Moving nova functional test def to in tree","message":"Moving nova functional test def to in tree\n\nA new functional job definition is added in the nova tree in\nIa684786d1622da7af31aa4479fc883a7c65848ff to make sure that the\nirrelevant-files is properly overridden. Therefore\nopenstack-tox-functional and functional-py35 is removed from the nova\npart of the project-config.\n\nDepends-On: Ia684786d1622da7af31aa4479fc883a7c65848ff\nChange-Id: I56d44f8dff41dbf3b2ff2382fa39b364f55f9a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ccbe79648af49a135da23af9c543f0876dd2e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ccbe79648af49a135da23af9c543f0876dd2e40"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d57cd43bf95f055320f365897511b285f80052c4","is_merged_in_target_branch":true,"change_id":"Ibdd11c4d461b8434bf90df70864c4fcdbbc2304e","change_number":539016,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
