)]}'
{"id":"openstack%2Fproject-config~568174","triplet_id":"openstack%2Fproject-config~master~I5d5624f670bd189eab18c444922220a12d1dc4a0","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I5d5624f670bd189eab18c444922220a12d1dc4a0","subject":"Retire kolla-kubernetes project - step 4 remove from infra system","status":"MERGED","created":"2018-05-14 02:53:59.000000000","updated":"2018-06-07 13:30:48.000000000","submitted":"2018-06-07 13:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"568174-1528378248731-c78f0fab","meta_rev_id":"94562b7c0c3cbae44c27caf1a05c589fa409fdd3","_number":568174,"virtual_id_number":568174,"owner":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2018-06-07 13:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"all":[{"value":1,"date":"2018-06-07 09:40:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":2,"date":"2018-06-07 09:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-06-07 09:40:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-06-07 13:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-06-07 13:21:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-14 09:07:03.000000000","updated_by":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"reviewer":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"state":"REVIEWER"},{"updated":"2018-05-17 17:07:33.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2018-06-07 09:42:45.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-06-07 13:21:42.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2018-06-07 13:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7cd4c497e44dca15cfcd8c4907c0a5de5bb0a7f1","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-05-14 02:53:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6221aac63c73ac28a5cd43e025fa452def3474b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-14 02:54:38.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"640d14aeab1eb1858a6fd697bb364e4de26a94d9","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-05-14 06:42:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ff66cb2e3772c772bf3ccad1dc72f42a6d91f4aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-05-14 06:45:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nAlso, remove jobs from openstack-zuul-jobs like:\n\nhttp://git.openstack.org/cgit/openstack-infra/openstack-zuul-jobs/tree/playbooks/legacy/periodic-kolla-kubernetes-deploy-centos-binary-2-ceph/run.yaml","accounts_in_message":[],"_revision_number":2},{"id":"17bb67f7cb9831ed828f64d889feccc9411a6e19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-14 06:45:53.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"6f62846674b8f6e5af1d97c92a79bafec42c1332","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-05-14 06:47:37.000000000","message":"Patch Set 2:\n\nAlso, remove the entry from requirements:\n\nhttp://git.openstack.org/cgit/openstack/requirements/tree/projects.txt#n72","accounts_in_message":[],"_revision_number":2},{"id":"b499a3936791a8f30b87a46d2ddba2954833f7dc","author":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"date":"2018-05-14 09:07:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"028fbc4b72d3a12f6add4a07a3f76e6c16d31ee1","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-05-17 06:46:11.000000000","message":"Patch Set 2:\n\nAndreas Jaeger wrote:\n\n\u003e Also, remove the entry from requirements:\n\u003e\n\u003e http://git.openstack.org/cgit/openstack/requirements/tree/projects.t\n\u003e xt#n72\n\ndone in another patch.","accounts_in_message":[],"_revision_number":2},{"id":"3c133bee25048017cd335124dc4ae521b2610ade","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-05-17 08:17:04.000000000","message":"Patch Set 2:\n\nJeffrey, please fix the projects.yaml file so that we can continue...","accounts_in_message":[],"_revision_number":2},{"id":"97d2b0418741bda7e8c7f8dcd6b69284c4a9c4a5","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-05-17 14:34:48.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c29f4e736052c6d9008b65bfb3d496b18bd8123a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-17 14:44:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/568174/3/check/openstack-zuul-jobs-linters/9910406/ : SUCCESS in 4m 17s\n- project-config-gerrit http://logs.openstack.org/74/568174/3/check/project-config-gerrit/67ec855/ : SUCCESS in 2m 46s\n- project-config-irc-access http://logs.openstack.org/74/568174/3/check/project-config-irc-access/f92536a/ : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":3},{"id":"b44ea5e5e87f6d5626da53e2b9e13fd9f04e3f23","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2018-05-17 17:07:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ed0fc64be97f821b20f6a5bb8c9753cd811c3985","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-05-17 17:08:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4edc477672d8cec76e152db2abf4ef65284679bb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-06-07 09:20:57.000000000","message":"Patch Set 3: -Code-Review Workflow-1\n\nPlease rebase so that we can merge this.\n\nAlso, abandon all open reviews, see https://review.openstack.org/#/q/project:openstack/kolla-kubernetes+is:open","accounts_in_message":[],"_revision_number":3},{"id":"823e2fe3c7ef656fd8c4d3cfbf41f39b43eb75bc","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-06-07 09:36:28.000000000","message":"Patch Set 3: Code-Review+1\n\nok, this patch is ready to go","accounts_in_message":[],"_revision_number":3},{"id":"d83db8235f7c13d69c182054ab5928f3aa1f87f4","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2018-06-07 09:40:02.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"60d05b2e53e1b86363f5119585c65ff903eb8964","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-06-07 09:42:45.000000000","message":"Patch Set 4: Code-Review+2\n\nAll looks fine...","accounts_in_message":[],"_revision_number":4},{"id":"6e994fdab8e404185343c125b0ff346ae27ddd95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 09:57:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/568174/4/check/openstack-zuul-jobs-linters/29615b9/ : SUCCESS in 4m 01s\n- project-config-gerrit http://logs.openstack.org/74/568174/4/check/project-config-gerrit/5fca187/ : SUCCESS in 3m 07s\n- project-config-irc-access http://logs.openstack.org/74/568174/4/check/project-config-irc-access/6d95d38/ : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":4},{"id":"e1fb1c696a550698bfdcf5117e44c86a89f7999a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-06-07 13:21:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b734412c985ef2d8769c899f31df0c4b43c4cc6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 13:22:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c8c7547afe45713994bba27fca985fe59bb8abb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 13:30:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/568174/4/gate/openstack-zuul-jobs-linters/36b716d/ : SUCCESS in 5m 31s\n- project-config-gerrit http://logs.openstack.org/74/568174/4/gate/project-config-gerrit/004dbc3/ : SUCCESS in 3m 28s\n- project-config-irc-access http://logs.openstack.org/74/568174/4/gate/project-config-irc-access/bdc375c/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"94562b7c0c3cbae44c27caf1a05c589fa409fdd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 13:30:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"458eb836834c501c086dd40ae3110cf8df078bc1","revisions":{"35ee4606676a8690fd63d9da9b64cc7809d6a807":{"kind":"REWORK","_number":1,"created":"2018-05-14 02:53:59.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/74/568174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/568174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/568174/1"}}},"commit":{"parents":[{"commit":"de2c3710fac4ce70ba0f326261bf5654d6c4d275","subject":"Retire kolla-kubernetes projects - step 2 end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de2c3710fac4ce70ba0f326261bf5654d6c4d275"}]}],"author":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-05-14 02:52:43.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-05-14 02:52:43.000000000","tz":480},"subject":"Retire kolla-kubernetes projects - step 4 remove from infra system","message":"Retire kolla-kubernetes projects - step 4 remove from infra system\n\nChange-Id: I5d5624f670bd189eab18c444922220a12d1dc4a0\nDepends-On: https://review.openstack.org/#/c/565221/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/35ee4606676a8690fd63d9da9b64cc7809d6a807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/35ee4606676a8690fd63d9da9b64cc7809d6a807"}]},"branch":"refs/heads/master"},"04dc07703579ef19c115c2572cc74d4522a8e600":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-05-14 06:42:12.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/74/568174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/568174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/568174/2"}}},"commit":{"parents":[{"commit":"de2c3710fac4ce70ba0f326261bf5654d6c4d275","subject":"Retire kolla-kubernetes projects - step 2 end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de2c3710fac4ce70ba0f326261bf5654d6c4d275"}]}],"author":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-05-14 02:52:43.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2018-05-14 06:42:08.000000000","tz":0},"subject":"Retire kolla-kubernetes project - step 4 remove from infra system","message":"Retire kolla-kubernetes project - step 4 remove from infra system\n\nChange-Id: I5d5624f670bd189eab18c444922220a12d1dc4a0\nDepends-On: https://review.openstack.org/#/c/565221/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04dc07703579ef19c115c2572cc74d4522a8e600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04dc07703579ef19c115c2572cc74d4522a8e600"}]},"branch":"refs/heads/master"},"b91fde0105bb3324a75befdc7f92fb361b8eba33":{"kind":"REWORK","_number":3,"created":"2018-05-17 14:34:48.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/74/568174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/568174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/568174/3"}}},"commit":{"parents":[{"commit":"de2c3710fac4ce70ba0f326261bf5654d6c4d275","subject":"Retire kolla-kubernetes projects - step 2 end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de2c3710fac4ce70ba0f326261bf5654d6c4d275"}]}],"author":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-05-14 02:52:43.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"jeffrey.zhang@99cloud.net","date":"2018-05-17 14:34:42.000000000","tz":0},"subject":"Retire kolla-kubernetes project - step 4 remove from infra system","message":"Retire kolla-kubernetes project - step 4 remove from infra system\n\nChange-Id: I5d5624f670bd189eab18c444922220a12d1dc4a0\nDepends-On: https://review.openstack.org/#/c/565221/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b91fde0105bb3324a75befdc7f92fb361b8eba33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b91fde0105bb3324a75befdc7f92fb361b8eba33"}]},"branch":"refs/heads/master"},"458eb836834c501c086dd40ae3110cf8df078bc1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-06-07 09:40:02.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/74/568174/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/568174/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/568174/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/568174/4"}}},"commit":{"parents":[{"commit":"93923644cf6106dad0361002a07a70cdbd98c10c","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93923644cf6106dad0361002a07a70cdbd98c10c"}]}],"author":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-05-14 02:52:43.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-06-07 09:39:53.000000000","tz":480},"subject":"Retire kolla-kubernetes project - step 4 remove from infra system","message":"Retire kolla-kubernetes project - step 4 remove from infra system\n\nChange-Id: I5d5624f670bd189eab18c444922220a12d1dc4a0\nDepends-On: https://review.openstack.org/#/c/565221/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/458eb836834c501c086dd40ae3110cf8df078bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/458eb836834c501c086dd40ae3110cf8df078bc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
