)]}'
{"id":"openstack%2Fproject-config~602011","triplet_id":"openstack%2Fproject-config~master~Ief8eb739ae0c5185c36dedf786b269b609de8209","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ief8eb739ae0c5185c36dedf786b269b609de8209","subject":"Follow devstack-plugin-ceph job rename","status":"ABANDONED","created":"2018-09-12 11:52:00.000000000","updated":"2018-09-24 06:52:29.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6881428749fcfe81e47662c9257dbadff408bf44","_number":602011,"virtual_id_number":602011,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-09-12 12:22:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2018-09-21 12:12:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-12 12:22:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-21 12:12:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"1300ba662acda88d1a5467e65621ee4ce895333e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-12 11:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbd2783f27d41faabd1ffc4567b375de71849dd4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-09-12 11:56:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nshould not be needed - we remove as part of python3-first all these jobs and move them in-repo where the change needs to happen.","accounts_in_message":[],"_revision_number":1},{"id":"1116a1c7e93751728ce06cd47006d11f82b1aa63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 12:03:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/11/602011/1/check/openstack-zuul-jobs-linters/f90fba1/ : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c3b2085312e876e9f912bb0fbb2884f09428d89e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-12 12:05:34.000000000","message":"Patch Set 1:\n\nyeah, all of them are taken care there but glance one is needed - https://review.openstack.org/#/c/602011/1/zuul.d/projects.yaml@5516","accounts_in_message":[],"_revision_number":1},{"id":"de0c404dca9329e01fa2d20c09fe7eb29d84ac37","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-12 12:11:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"735ddee103eda9921e08191614368801ca7a74fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-12 12:22:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/11/602011/2/check/openstack-zuul-jobs-linters/afbcc3d/ : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"af998fdefe5aefacb6ac48cfa22f76cf6f40b855","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-09-13 10:44:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cdc42e931a3b394e0aa9510a31b5c436d070f0ca","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-09-21 12:12:59.000000000","message":"Patch Set 2: Code-Review-1\n\nhandled via https://review.openstack.org/593525 and https://review.openstack.org/#/c/603197/","accounts_in_message":[],"_revision_number":2},{"id":"6881428749fcfe81e47662c9257dbadff408bf44","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-09-24 06:52:29.000000000","message":"Abandoned\n\nMoved in-tree and handled there","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a3d2bc686c694436716b00d68bc6859134f35904","revisions":{"7d23a1c4b792b42232595eda54c41ece45461745":{"kind":"REWORK","_number":1,"created":"2018-09-12 11:52:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/602011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/602011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/602011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/602011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/602011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/602011/1"}}},"commit":{"parents":[{"commit":"469aa526cad8331dede7947a7ca90dc8e7086a68","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/469aa526cad8331dede7947a7ca90dc8e7086a68"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-12 11:51:07.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-12 11:51:07.000000000","tz":0},"subject":"Follow devstack-plugin-ceph job rename","message":"Follow devstack-plugin-ceph job rename\n\nThe devstack-plugin-ceph jobs have migrated to zuul v3\n\u0026 in-repo and renamed[1], this commit use the new job.\n\nPatch removing old job from openstack-infra/openstack-zuul-jobs[2]\n[1] https://review.openstack.org/543048\n[2] https://review.openstack.org/#/c/601316/\n\nChange-Id: Ief8eb739ae0c5185c36dedf786b269b609de8209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d23a1c4b792b42232595eda54c41ece45461745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d23a1c4b792b42232595eda54c41ece45461745"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"469aa526cad8331dede7947a7ca90dc8e7086a68","is_merged_in_target_branch":true,"change_id":"I03824f6184ee03f072058fe49831c7032cdf65e1","change_number":601916,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a3d2bc686c694436716b00d68bc6859134f35904":{"kind":"REWORK","_number":2,"created":"2018-09-12 12:11:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/602011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/602011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/602011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/602011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/602011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/602011/2"}}},"commit":{"parents":[{"commit":"469aa526cad8331dede7947a7ca90dc8e7086a68","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/469aa526cad8331dede7947a7ca90dc8e7086a68"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-12 12:09:13.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-12 12:09:13.000000000","tz":0},"subject":"Follow devstack-plugin-ceph job rename","message":"Follow devstack-plugin-ceph job rename\n\nThe devstack-plugin-ceph jobs have migrated to zuul v3\n\u0026 in-repo and renamed[1], this commit use the new job.\n\nThis patch rename job for openstack/glance only as other\nplace is taken care in https://review.openstack.org/#/c/601032/\nPatch removing old job from openstack-infra/openstack-zuul-jobs[2]\n[1] https://review.openstack.org/543048\n[2] https://review.openstack.org/#/c/601316/\n\nChange-Id: Ief8eb739ae0c5185c36dedf786b269b609de8209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3d2bc686c694436716b00d68bc6859134f35904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3d2bc686c694436716b00d68bc6859134f35904"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"469aa526cad8331dede7947a7ca90dc8e7086a68","is_merged_in_target_branch":true,"change_id":"I03824f6184ee03f072058fe49831c7032cdf65e1","change_number":601916,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
