)]}'
{"id":"openstack%2Fproject-config~605026","triplet_id":"openstack%2Fproject-config~master~I9d3aa05a477dc230d66b58fee711667fd1faf07b","project":"openstack/project-config","branch":"master","topic":"graphana_fix_kolla","hashtags":[],"change_id":"I9d3aa05a477dc230d66b58fee711667fd1faf07b","subject":"Fix not working kolla graphs","status":"MERGED","created":"2018-09-25 09:57:29.000000000","updated":"2018-09-26 03:54:07.000000000","submitted":"2018-09-26 03:54:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"605026-1537934047096-71773c33","meta_rev_id":"c51fbb8873ebe2eddca63b427584827d5e9e7a92","_number":605026,"virtual_id_number":605026,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2018-09-26 03:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2018-09-25 10:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-09-26 03:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-26 03:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-25 10:43:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-09-26 03:37:58.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2018-09-26 03:54:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e66a9a99f26a02c6fa710d91831b7704440573d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2018-09-25 09:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"793ead5d57ecf6743f388eceb0fffdd1b43c28b7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-09-25 10:43:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f09aae0072a34e7f203f9bdd0eb569eb196cd78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-25 12:07:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/26/605026/1/check/openstack-zuul-jobs-linters/5b8f27a/ : SUCCESS in 6m 17s\n- project-config-grafyaml http://logs.openstack.org/26/605026/1/check/project-config-grafyaml/67753db/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d95fe918ef2b86585e2ba3ab665f98796be83a0b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2018-09-26 03:37:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8e9c6e399fab51e4591eb8859ef7efa92e7fc77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-26 03:38:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04f6fc3867eefc11875c5795afae6082f77869ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-26 03:54:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/26/605026/1/gate/openstack-zuul-jobs-linters/95be1c3/ : SUCCESS in 6m 24s\n- project-config-grafyaml http://logs.openstack.org/26/605026/1/gate/project-config-grafyaml/7c1051b/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1c256c4e7d338701e1c72822f5261e5d2f7d9aa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-26 03:54:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c17cd51a1551ef21befcff6d2331750154608b6","revisions":{"0c17cd51a1551ef21befcff6d2331750154608b6":{"kind":"REWORK","_number":1,"created":"2018-09-25 09:57:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/605026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/605026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/605026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/605026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/605026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/605026/1"}}},"commit":{"parents":[{"commit":"9ae88de3e258e1ee21ebb117fcaa0eb2488cc2e7","subject":"Merge \"Add cherrypy to third-party ci\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ae88de3e258e1ee21ebb117fcaa0eb2488cc2e7"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2018-09-25 09:55:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2018-09-25 09:57:26.000000000","tz":120},"subject":"Fix not working kolla graphs","message":"Fix not working kolla graphs\n\n- Fixed stats path for existing kolla build graphs\n- Removed deployment failures graphs - will be moved to kolla-ansible dashboard\n\nChange-Id: I9d3aa05a477dc230d66b58fee711667fd1faf07b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c17cd51a1551ef21befcff6d2331750154608b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c17cd51a1551ef21befcff6d2331750154608b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
