)]}'
{"id":"openstack%2Fproject-config~615501","triplet_id":"openstack%2Fproject-config~master~I5cba38e1a2a6c01e692a542efad6b4df77cd0751","project":"openstack/project-config","branch":"master","topic":"publish-static","hashtags":[],"change_id":"I5cba38e1a2a6c01e692a542efad6b4df77cd0751","subject":"Update static.o.o publishing","status":"MERGED","created":"2018-11-05 08:46:10.000000000","updated":"2018-11-05 20:25:53.000000000","submitted":"2018-11-05 20:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"615501-1541449553336-bba6691b","meta_rev_id":"a1634c8ca321da92572b3ccdb691c88e15e0a117","_number":615501,"virtual_id_number":615501,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2018-11-05 20:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2018-11-05 20:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-11-05 20:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-11-05 20:18:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-05 20:15:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2018-11-05 20:18:05.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2018-11-05 20:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de82e2943fb028f814c3b14a79a8edf10d321a4d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-05 08:46:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01e49f863ca752a24a1b87babff342fda5d54d00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 08:49:24.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e46c52ab27404c0a6f0d0229a412b86a36b9fb4c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-05 08:49:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b7ef0587d33dcea7fe65fba1c2b30d94b2bb546","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 08:58:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/01/615501/1/check/openstack-zuul-jobs-linters/18c0b3a/ : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0b83bb5bebfcbdd4155670b7d742c4886d580993","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2018-11-05 18:59:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f798c0ec4b08aff6557701de0faa51492440ab97","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2018-11-05 19:28:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2e0512edd996feecb78e5295272be45710dc9f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 19:28:19.000000000","message":"Patch Set 1: Verified-2\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack-infra/project-config on branch master.  The error was:\n\n  The project template \"build-openstack-sphinx-pti\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/governance-website\n      templates:\n        - build-openstack-sphinx-pti\n      post:\n        jobs:\n          - publish-governance-website\n\n  in \"openstack-infra/project-config/zuul.d/projects.yaml@master\", line 4186, column 3","accounts_in_message":[],"_revision_number":1},{"id":"41ac3ec62174823556f293aa215a18c9d4c7261a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-05 19:41:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2654d5eb5e890c7afc141bd74122a58b883372ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 19:41:39.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/election on branch master.  The error was:\n\n  Job publish-static not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - build-openstack-sphinx-docs\n        - openstack-python35-jobs\n        - openstack-python36-jobs\n      check:\n        jobs:\n          - openstack-tox-py27\n          - openstack-tox-linters\n          - election-tox-ci-checks\n      gate:\n        jobs:\n          - openstack-tox-py27\n          - openstack-tox-linters\n          - election-tox-ci-checks\n      post:\n        jobs:\n          - publish-static\n\n  in \"openstack/election/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"5a091c61da7582a2d3004938cf08421b6c384c35","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-05 19:49:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab5d852839375fb11de5b79bbb5e287cf5054320","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 20:08:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/01/615501/3/check/openstack-zuul-jobs-linters/75e0f6f/ : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":3},{"id":"6d5c3e1e38bee294a1aee89f3355a610eee05442","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2018-11-05 20:15:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ef5a7c8da95b5a75a7b4fd2c7fc2754d3e6b0c58","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2018-11-05 20:18:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b16f3fa8178b1e00152ac1876e11e39da17c03ce","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2018-11-05 20:18:05.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0c8d8b4095c809da91f81fa52c111d66ad3b4525","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 20:18:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"277f579bbd4a1c412c3f23267eb237b6da6c0580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 20:25:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/01/615501/3/gate/openstack-zuul-jobs-linters/78047c4/ : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":3},{"id":"eb20807e5e1d5ba86dd292febca35701a4aaf98f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 20:25:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"84b80f2259cb2726239cbd15324ab935565b5a50","revisions":{"4353641b94a036d0591788ca132bb593032abee7":{"kind":"REWORK","_number":1,"created":"2018-11-05 08:46:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/615501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/615501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/615501/1"}}},"commit":{"parents":[{"commit":"edb6b70edc1c5ca5545aed6dada5ac7c36bae5b3","subject":"Merge \"Remove python3 release jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edb6b70edc1c5ca5545aed6dada5ac7c36bae5b3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 08:40:47.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 08:40:47.000000000","tz":60},"subject":"Update static.o.o publishing","message":"Update static.o.o publishing\n\nWe can now use the \u0027tox -e docs\u0027 publishing for all repos, update the\nremaining static publishing docs.\n\nRemove base-publish-static, it is not needed anymore.\n\nUpdate governance-website and transparent-policy to use new docs-pti\njob, the tox.ini files for these repos handle \"tox -e docs\".\n\nChange-Id: I5cba38e1a2a6c01e692a542efad6b4df77cd0751\nDepends-On: https://review.openstack.org/615498\nDepends-On: https://review.openstack.org/615497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4353641b94a036d0591788ca132bb593032abee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4353641b94a036d0591788ca132bb593032abee7"}]},"branch":"refs/heads/master"},"7c99d72f2fee66692245b62917d8c6c78bb86cde":{"kind":"REWORK","_number":2,"created":"2018-11-05 19:41:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/615501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/615501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/615501/2"}}},"commit":{"parents":[{"commit":"edb6b70edc1c5ca5545aed6dada5ac7c36bae5b3","subject":"Merge \"Remove python3 release jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edb6b70edc1c5ca5545aed6dada5ac7c36bae5b3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 08:40:47.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 19:40:30.000000000","tz":60},"subject":"Update static.o.o publishing","message":"Update static.o.o publishing\n\nWe can now use the \u0027tox -e docs\u0027 publishing for all repos, update the\nremaining static publishing docs.\n\nRemove base-publish-static, it is not needed anymore.\n\nUpdate governance-website and transparent-policy to use new docs-pti\njob, the tox.ini files for these repos handle \"tox -e docs\".\n\nChange-Id: I5cba38e1a2a6c01e692a542efad6b4df77cd0751\nDepends-On: https://review.openstack.org/615498\nDepends-On: https://review.openstack.org/615497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c99d72f2fee66692245b62917d8c6c78bb86cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c99d72f2fee66692245b62917d8c6c78bb86cde"}]},"branch":"refs/heads/master"},"84b80f2259cb2726239cbd15324ab935565b5a50":{"kind":"REWORK","_number":3,"created":"2018-11-05 19:49:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/615501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/615501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/615501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/615501/3"}}},"commit":{"parents":[{"commit":"edb6b70edc1c5ca5545aed6dada5ac7c36bae5b3","subject":"Merge \"Remove python3 release jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edb6b70edc1c5ca5545aed6dada5ac7c36bae5b3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 08:40:47.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 19:48:56.000000000","tz":60},"subject":"Update static.o.o publishing","message":"Update static.o.o publishing\n\nWe can now use the \u0027tox -e docs\u0027 publishing for all repos, update the\nremaining static publishing docs.\n\nChange use of publish-static to new defined job publish-tox-docs-static,\nwe can remove publish-static once all users are updated but need both\njobs for now for the transition.\n\nUpdate governance-website and transparent-policy to use new docs-pti\njob, the tox.ini files for these repos handle \"tox -e docs\".\n\nChange-Id: I5cba38e1a2a6c01e692a542efad6b4df77cd0751\nDepends-On: https://review.openstack.org/615498\nDepends-On: https://review.openstack.org/615497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84b80f2259cb2726239cbd15324ab935565b5a50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84b80f2259cb2726239cbd15324ab935565b5a50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
