)]}'
{"id":"openstack%2Fproject-config~615637","triplet_id":"openstack%2Fproject-config~master~I4751100208008ee912df20e8df4f253631183e22","project":"openstack/project-config","branch":"master","topic":"publish-static","hashtags":[],"change_id":"I4751100208008ee912df20e8df4f253631183e22","subject":"Remove publish-static","status":"MERGED","created":"2018-11-05 19:55:46.000000000","updated":"2018-11-07 07:18:49.000000000","submitted":"2018-11-07 07:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"615637-1541575129625-b083b179","meta_rev_id":"5efaeb095f4484269ac06323f9a979d9935b5b49","_number":615637,"virtual_id_number":615637,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2018-11-06 19:51:56.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2018-11-07 07:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2018-11-05 20:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-11-07 07:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-11-07 07:07:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-05 20:15:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2018-11-07 07:07:47.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2018-11-07 07:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a78f5b4d267743e1286245a92a01ac1ccb2a874c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-05 19:55:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf94e7f5009fc66095755c1494ebb993498b4c17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 19:56:13.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/election on branch master.  The error was:\n\n  Job publish-tox-docs-static not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - build-openstack-docs-pti\n        - openstack-python35-jobs\n        - openstack-python36-jobs\n      check:\n        jobs:\n          - openstack-tox-py27\n          - openstack-tox-linters\n          - election-tox-ci-checks\n      gate:\n        jobs:\n          - openstack-tox-py27\n          - openstack-tox-linters\n          - election-tox-ci-checks\n      post:\n        jobs:\n          - publish-tox-docs-static\n\n  in \"openstack/election/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"124a8f90cc9e96a02e14cada47b6394a0c90fc34","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-05 19:59:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bfbc05f94c0dab152771ec3ba40819093d4d660c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 20:00:04.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change to a config project.\n\nThe syntax of the configuration in this change has been verified to be\ncorrect once the config project change upon which it depends is\nmerged, but it can not be used until that occurs.","accounts_in_message":[],"_revision_number":2},{"id":"90c336d2a930aa1d93074b85e00a2ca790e0b7ef","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2018-11-05 20:15:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d8dfb4ffb19a229de40f6dc17b968559f02d5e5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-06 19:51:56.000000000","message":"Patch Set 2:\n\nrecheck - dependency merged","accounts_in_message":[],"_revision_number":2},{"id":"e48ad09da15f55667a12179a97bf33150ba24d34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-06 19:59:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/37/615637/2/check/openstack-zuul-jobs-linters/361ae7a/ : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"492a0101fea0ea7c744e4d69dec27c0dce988fb8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2018-11-07 07:07:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c52d53f1204b66a1eb9d447f1dbec17ee351b994","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 07:08:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb9c472c2f7eb5e106ea31ce3637c821b983d635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 07:18:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/37/615637/2/gate/openstack-zuul-jobs-linters/2147247/ : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ec26018c9d418c4cf96f81c8e26c765a845d0b4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 07:18:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bbec3f3aba6bb3b0394de546e0a0bf97c5677085","revisions":{"dd4bddfc3d856c05921c6e75bf6475b5ce1c670e":{"kind":"REWORK","_number":1,"created":"2018-11-05 19:55:46.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/615637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/615637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/615637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/615637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/615637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/615637/1"}}},"commit":{"parents":[{"commit":"84b80f2259cb2726239cbd15324ab935565b5a50","subject":"Update static.o.o publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84b80f2259cb2726239cbd15324ab935565b5a50"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 19:54:10.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 19:54:10.000000000","tz":60},"subject":"Remove publish-static","message":"Remove publish-static\n\npublish-static and base-publish-static are not used anymore, remove\nthem.\n\nChange-Id: I4751100208008ee912df20e8df4f253631183e22\nDepends-On: https://review.openstack.org/615636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd4bddfc3d856c05921c6e75bf6475b5ce1c670e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd4bddfc3d856c05921c6e75bf6475b5ce1c670e"}]},"branch":"refs/heads/master"},"bbec3f3aba6bb3b0394de546e0a0bf97c5677085":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-11-05 19:59:41.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/615637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/615637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/615637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/615637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/615637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/615637/2"}}},"commit":{"parents":[{"commit":"84b80f2259cb2726239cbd15324ab935565b5a50","subject":"Update static.o.o publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84b80f2259cb2726239cbd15324ab935565b5a50"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 19:54:10.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2018-11-05 19:59:11.000000000","tz":60},"subject":"Remove publish-static","message":"Remove publish-static\n\npublish-static and base-publish-static are not used anymore, remove\nthem.\n\nDepends-On: https://review.openstack.org/615636\nDepends-On: https://review.openstack.org/615639\nChange-Id: I4751100208008ee912df20e8df4f253631183e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbec3f3aba6bb3b0394de546e0a0bf97c5677085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbec3f3aba6bb3b0394de546e0a0bf97c5677085"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
