)]}'
{"id":"openstack%2Fproject-config~616676","triplet_id":"openstack%2Fproject-config~master~I94daa02a0380a765490b8e2b9d0276bcdba26103","project":"openstack/project-config","branch":"master","topic":"twine-for-release-test","hashtags":[],"change_id":"I94daa02a0380a765490b8e2b9d0276bcdba26103","subject":"ensure twine is installed with python 3","status":"MERGED","created":"2018-11-08 19:11:16.000000000","updated":"2018-11-09 14:16:11.000000000","submitted":"2018-11-09 14:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"616676-1541772971279-eb36e3b5","meta_rev_id":"6b0c12cf23b2f26669d61d0d7a1dc68a38418eef","_number":616676,"virtual_id_number":616676,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2018-11-09 14:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26109,"name":"Aakarsh","email":"aakarsh.g2012@gmail.com","username":"agopi"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2018-11-08 20:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-11-08 20:57:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26109,"name":"Aakarsh","email":"aakarsh.g2012@gmail.com","username":"agopi"},{"value":2,"date":"2018-11-08 21:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2018-11-09 14:05:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26109,"name":"Aakarsh","email":"aakarsh.g2012@gmail.com","username":"agopi"},{"value":1,"date":"2018-11-08 21:04:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26109,"name":"Aakarsh","email":"aakarsh.g2012@gmail.com","username":"agopi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-08 19:13:31.000000000","updated_by":{"_account_id":26109,"name":"Aakarsh","email":"aakarsh.g2012@gmail.com","username":"agopi"},"reviewer":{"_account_id":26109,"name":"Aakarsh","email":"aakarsh.g2012@gmail.com","username":"agopi"},"state":"REVIEWER"},{"updated":"2018-11-08 21:04:26.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2018-11-09 14:05:09.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-11-09 14:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"599a8eb0344e4f120228a12d6e6caa2c6d847d83","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-11-08 19:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"809b14d4b03051a0d92d62ca582d3db65ad3cf0f","author":{"_account_id":26109,"name":"Aakarsh","email":"aakarsh.g2012@gmail.com","username":"agopi"},"date":"2018-11-08 19:13:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"169b0166d11d539d9764a4ab3e18de140be7cb06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 19:18:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/76/616676/1/check/openstack-zuul-jobs-linters/839e346/ : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"12b90a987581206b8a731a6cd489d346d66aedae","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-08 19:47:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70bb57c545ea41a3d202834276831f2e409c32c5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-08 20:57:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d632691f9319c36a9aac8b85d3630d7a87065e12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 21:03:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/76/616676/2/check/openstack-zuul-jobs-linters/34c6b4c/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"54985544f50a66962eb4315b86bed1dcff168495","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2018-11-08 21:04:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7269a3073bc1cc6d9d17dd21d255aed81e089c6e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-09 14:05:09.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"796fbf0437ef08af185512bfffefb0ec7cc37167","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 14:05:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"99e9cc536df583830c6a06dc15e78d24883ef2f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 14:16:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/76/616676/2/gate/openstack-zuul-jobs-linters/fa5bb9e/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"095dbd8ab21b8919888d72a03136f3e901c681c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 14:16:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f3b36607738be78985982c3dc935836e9a1814d","revisions":{"7fc34c4ec63d28d4f7f9b58e77fe81629cee5dbd":{"kind":"REWORK","_number":1,"created":"2018-11-08 19:11:16.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/76/616676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/616676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/616676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/616676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/616676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/616676/1"}}},"commit":{"parents":[{"commit":"79cc1ab2c3aaa8916b7392bd7dfcf75d3daeef09","subject":"remove setup.py check from check-python-release role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/79cc1ab2c3aaa8916b7392bd7dfcf75d3daeef09"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-11-08 19:10:36.000000000","tz":-300},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-11-08 19:10:36.000000000","tz":-300},"subject":"ensure twine is installed with python 3","message":"ensure twine is installed with python 3\n\nSet twine_python to python3 to ensure that we use python3 to install\ntwine.\n\nChange-Id: I94daa02a0380a765490b8e2b9d0276bcdba26103\nDepends-On: https://review.openstack.org/616673\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fc34c4ec63d28d4f7f9b58e77fe81629cee5dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fc34c4ec63d28d4f7f9b58e77fe81629cee5dbd"}]},"branch":"refs/heads/master"},"4f3b36607738be78985982c3dc935836e9a1814d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-11-08 20:57:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/76/616676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/616676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/616676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/616676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/616676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/616676/2"}}},"commit":{"parents":[{"commit":"53f1b26469489571a6642b57941564d1342912c0","subject":"Remove check-python-release role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53f1b26469489571a6642b57941564d1342912c0"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-11-08 19:10:36.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2018-11-08 20:57:28.000000000","tz":0},"subject":"ensure twine is installed with python 3","message":"ensure twine is installed with python 3\n\nSet twine_python to python3 to ensure that we use python3 to install\ntwine.\n\nChange-Id: I94daa02a0380a765490b8e2b9d0276bcdba26103\nDepends-On: https://review.openstack.org/616673\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f3b36607738be78985982c3dc935836e9a1814d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f3b36607738be78985982c3dc935836e9a1814d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
