)]}'
{"id":"openstack%2Fproject-config~618511","triplet_id":"openstack%2Fproject-config~master~Ifcb05fc5c666e85f3c483f8d542dd1485cd05a11","project":"openstack/project-config","branch":"master","topic":"rebase-zuul","hashtags":[],"change_id":"Ifcb05fc5c666e85f3c483f8d542dd1485cd05a11","subject":"Add openstack/tenks Ansible dependencies","status":"MERGED","created":"2018-11-16 11:03:50.000000000","updated":"2019-05-06 06:39:15.000000000","submitted":"2019-05-06 06:39:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"618511-1557124755603-0b0d46a4","meta_rev_id":"d2329d8116b6fbbb8adeb7283d0f43bcf005f1d7","_number":618511,"virtual_id_number":618511,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-01 10:42:29.000000000","_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-06 06:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-05-03 07:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-06 06:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-05-03 07:15:54.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-06 06:31:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-19 17:22:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-03 07:15:54.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-05-03 07:15:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-06 06:31:39.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-05-06 06:39:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e76691544d1453ae2ac150f98a0a9f246bbce264","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2018-11-16 11:03:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4432544041ee650a40a3672c20e068835fcf0b4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-16 11:13:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/11/618511/1/check/openstack-zuul-jobs-linters/720970a/ : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cad60aa6c3d7ab0790e8c329ac82b984225d295b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-19 17:05:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de3f10e6d6995e3d0f8eb3599af16f5c79fab9f3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2018-11-19 17:08:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"07d4fa3592a454f09bc62c17fac23a4632d59ef9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-19 17:22:05.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI think we also need to add jobs for these projects in zuul.d/projects.yaml. See the first few jobs in that file for examples of using the third-party-check pipeline.","accounts_in_message":[],"_revision_number":2},{"id":"5945899b8d5be98931a12689e2c752664ef8fa1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 17:47:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/11/618511/2/check/openstack-zuul-jobs-linters/7e7cf3e/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9e152b6c039044974d42a9daab7d781cf7daf088","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2018-11-19 18:52:01.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a13e4c4509a23f38ac6a98b404bc5cdf25cd1265","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-19 18:52:40.000000000","message":"Patch Set 2:\n\nWill, could you show me a change that is failing - and that this one should fix, please?","accounts_in_message":[],"_revision_number":2},{"id":"def37b05162158d809f802cf0340d770f4b92aeb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2018-11-20 10:12:57.000000000","message":"Patch Set 2:\n\nSure, here is an example of me testing an upstream change:\n\nhttps://review.openstack.org/#/c/618119/\n\nbut it would have been nicer if we could use:\n\n   Depends-On: https://github.com/stackhpc/ansible-role-libvirt-host/pull/8\n   Depends-On: https://github.com/stackhpc/ansible-role-libvirt-vm/pull/16\n\nin the commit message. \n\nI then had to depend on this branch that would not merged:\n\nhttps://review.openstack.org/#/c/618245/\n\nUsing the \"Depends-On\" syntax, I could avoid these problems. What do you think?","accounts_in_message":[],"_revision_number":2},{"id":"8fdf27475e12ab6a914a10ae970e86772d50f313","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-20 10:43:52.000000000","message":"Patch Set 2:\n\nSo, you want to do something like:\nhttp://git.openstack.org/cgit/openstack/openstack-ansible/tree/zuul.d/jobs.yaml#n55\n\nThen indeed this change is all that\u0027s needed here. It\u0027s worth adding a comment on where those are needed.\n\nIn your project you would need to update your jobs to use \"required-projects\" for the job definitions so that the job always tests with that.\n\nSo, that needs an initial change to all of your jobs as well. Could you create that one as depends-on this one to show what you do? The depends-on will not work since this repos is a trusted repo - but should help us understand that this will work...","accounts_in_message":[],"_revision_number":2},{"id":"9e811db6212cc41e9bde7a654405f8749b852978","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-28 18:10:42.000000000","message":"Patch Set 2:\n\nWill, do you want to continue this one - or abandon it? I\u0027m waiting for a reply to my last comment and this needs rebasing as well.","accounts_in_message":[],"_revision_number":2},{"id":"37b656463abf5658799824c142e0aca6f870d7d9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-30 16:45:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8efbca4e2d6d937911ecf6f4a180f24777ee1131","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-30 16:46:10.000000000","message":"Patch Set 3:\n\nSorry Andreas, I\u0027d forgotten about this one. I will try and knock up the change you suggested.","accounts_in_message":[],"_revision_number":3},{"id":"0f83f444426d4136d1464a9d900139611d791c70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 17:01:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/11/618511/3/check/openstack-zuul-jobs-linters/4d7f4b8/ : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":3},{"id":"70fc70731d5c87005489556d9d3cef6b5d849ee1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-05-01 10:42:29.000000000","message":"Patch Set 3:\n\nHere is how we would use such a change: https://review.opendev.org/#/c/656627","accounts_in_message":[],"_revision_number":3},{"id":"1efc3d02bc81a3ee734d0ad90c8460a7685c5707","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-03 07:15:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9530950cfa0f90617377dd4461ce761ce3690d47","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-06 06:31:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c68eb428bccb9ff01483569bb58d9dd71d77e5da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:31:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"161713889115d6cbf2e65bedc2087744bbeb3a74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:39:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/11/618511/3/gate/openstack-zuul-jobs-linters/e614ab1/ : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":3},{"id":"8dfd4c267e1eb8d92539ba113dd68a5f4a96997d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:39:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5b6211a819b9c567a90da504b6297a5f7dadfe73","revisions":{"ad05c2a1e879923e3090e1cfb0e535ec0c673e0e":{"kind":"REWORK","_number":1,"created":"2018-11-16 11:03:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/11/618511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/618511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/618511/1"}}},"commit":{"parents":[{"commit":"fd9ccd0bb4012f03569dc74dcbe04b26504b5fd0","subject":"Merge \"Set translate option for contributor-guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd9ccd0bb4012f03569dc74dcbe04b26504b5fd0"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2018-11-16 10:58:58.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2018-11-16 11:04:57.000000000","tz":0},"subject":"Add openstack/tenks Ansible dependencies","message":"Add openstack/tenks Ansible dependencies\n\nIt would be useful to if would depend on pull requests for these\nprojects so that we can ensure they don\u0027t break tenks.\n\nChange-Id: Ifcb05fc5c666e85f3c483f8d542dd1485cd05a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad05c2a1e879923e3090e1cfb0e535ec0c673e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad05c2a1e879923e3090e1cfb0e535ec0c673e0e"}]},"branch":"refs/heads/master"},"8a93a512149564bad8c90ba498c19c630b084614":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-11-19 17:08:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/11/618511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/618511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/618511/2"}}},"commit":{"parents":[{"commit":"fd9ccd0bb4012f03569dc74dcbe04b26504b5fd0","subject":"Merge \"Set translate option for contributor-guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd9ccd0bb4012f03569dc74dcbe04b26504b5fd0"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2018-11-16 10:58:58.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2018-11-19 17:07:51.000000000","tz":0},"subject":"Add openstack/tenks Ansible dependencies","message":"Add openstack/tenks Ansible dependencies\n\nIt would be useful to if could depend on pull requests for these\nprojects so that we can ensure they don\u0027t break tenks.\n\nChange-Id: Ifcb05fc5c666e85f3c483f8d542dd1485cd05a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a93a512149564bad8c90ba498c19c630b084614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a93a512149564bad8c90ba498c19c630b084614"}]},"branch":"refs/heads/master"},"5b6211a819b9c567a90da504b6297a5f7dadfe73":{"kind":"REWORK","_number":3,"created":"2019-04-30 16:45:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/11/618511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/618511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/618511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/618511/3"}}},"commit":{"parents":[{"commit":"139c38acdc088eb9d3705d1440ef6b5280319936","subject":"Merge \"Stop master translations for tripleo-ui\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/139c38acdc088eb9d3705d1440ef6b5280319936"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2018-11-16 10:58:58.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-30 16:34:48.000000000","tz":60},"subject":"Add openstack/tenks Ansible dependencies","message":"Add openstack/tenks Ansible dependencies\n\nIt would be useful if we could depend on pull requests for these\nprojects so that we can ensure they don\u0027t break tenks.\n\nChange-Id: Ifcb05fc5c666e85f3c483f8d542dd1485cd05a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b6211a819b9c567a90da504b6297a5f7dadfe73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b6211a819b9c567a90da504b6297a5f7dadfe73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
