)]}'
{"id":"openstack%2Fproject-config~627573","triplet_id":"openstack%2Fproject-config~master~Iec643ec2af8c6c1567653608ef7851a5ac6e414a","project":"openstack/project-config","branch":"master","topic":"forge-publish","hashtags":[],"change_id":"Iec643ec2af8c6c1567653608ef7851a5ac6e414a","subject":"Add puppetforge publish and add to puppet release","status":"MERGED","created":"2018-12-27 22:41:04.000000000","updated":"2019-01-14 18:36:53.000000000","submitted":"2019-01-14 18:36:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"627573-1547491013217-d1a6a7d1","meta_rev_id":"58b0e8436e7a8df584acd228ffeb6a9bc3a15c87","_number":627573,"virtual_id_number":627573,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-01-14 18:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-01-14 16:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-01-14 18:21:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-01-14 18:21:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-01-14 17:15:36.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-12 15:53:43.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-01-14 16:54:51.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-01-14 18:21:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-01-14 18:36:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b29fcbd3dd8a5b85782e7c21c70454e4fc82fec","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-12-27 22:41:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4fe9b8d01e6fcfdd8e7b9f0813d69f3171bc1eb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-12-27 22:41:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"871528a92088706ddb16b29652d21688b8f39c42","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-12-27 22:41:59.000000000","message":"Patch Set 2: Workflow-1\n\nnot done yet, need to find what the name of the secret is for the \"openstack\" forge account","accounts_in_message":[],"_revision_number":2},{"id":"24ff0e264e7c8537c44f20aa00e472c95459e001","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 22:47:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/627573/2/check/openstack-zuul-jobs-linters/9783e65/ : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"377b00f8bbb63fe2e355c3ffd3f0c04ac97646ad","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-12-27 23:10:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"175bf18a5b51a0b2fad46f7c628e88dbc6c3bffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-27 23:19:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/627573/2/check/openstack-zuul-jobs-linters/d42fedd/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":2},{"id":"548c74d28aa22fcabad82671d2d74c6e58189c1e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-07 15:09:09.000000000","message":"Topic changed from puppetforge to forge-publish","accounts_in_message":[],"_revision_number":2},{"id":"2db5bd5e0a0d8fcf58ecb52e581e9ae6edf3a77f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-01-12 15:53:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1566f1f769dd4cbe81008ec4352067478b5d5e01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-12 16:01:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/627573/3/check/openstack-zuul-jobs-linters/9f447ae/ : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a621fff48f17b33692c14abf35e2b33c8ccbe52a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-14 08:34:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d412c7b4b73f5e01fed5e6b2409d7362b45262d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 08:41:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/627573/4/check/openstack-zuul-jobs-linters/e27685c/ : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":4},{"id":"72056a8c9ef77996b973ed5a5b4292ffe22bb79e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-01-14 10:03:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9dfae1d07102abf5200f690e6aa157e72b493e61","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-14 10:59:55.000000000","message":"Patch Set 4: Workflow-1\n\nWait for review on https://review.openstack.org/#/c/627553/ to clear and new changes to be made, need to update this patch as well.","accounts_in_message":[],"_revision_number":4},{"id":"03f2f4f204db5ed7564dcfcc1f34f017422c8024","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-14 11:32:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d14e670f1dc24fb84e624ca0ad59327f91343f74","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-14 11:33:12.000000000","message":"Patch Set 5: Workflow-1\n\nWait for [1] to be approved.","accounts_in_message":[],"_revision_number":5},{"id":"7d2ba0328632e507729dada9a4864e9e10e4ed99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 11:41:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/627573/5/check/openstack-zuul-jobs-linters/5b4a4a2/ : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":5},{"id":"93d7eaed073ad4c92fe49205764173a07069278b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-01-14 16:54:09.000000000","message":"Patch Set 5:\n\n\u003e Wait for [1] to be approved.\n\nNote the depends on will enforce this.","accounts_in_message":[],"_revision_number":5},{"id":"7bd377a58c65a874120af2428957c99a9161996d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-01-14 16:54:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"87b0f15c216f5bcb6eec5584cc3c9a4226d97625","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-01-14 17:15:36.000000000","message":"Patch Set 5: -Workflow\n\nI even missed that I had that, thanks.","accounts_in_message":[],"_revision_number":5},{"id":"4c75594040341517616f42fa0f92fc97fa54e551","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-01-14 18:21:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"729a8ebc6910f62b35a448ac7b360e2c17e7a100","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 18:21:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b7d7d9909bb9ddc0758244f6a118499da0675990","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 18:36:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/73/627573/5/gate/openstack-zuul-jobs-linters/c9083ef/ : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":5},{"id":"58b0e8436e7a8df584acd228ffeb6a9bc3a15c87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 18:36:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"afbc57f93b3b84dab502b1fa85d5dc40294bc08e","revisions":{"5f33dae985f9c58bba6bb3f9e39d9e4ec33cc9af":{"kind":"REWORK","_number":1,"created":"2018-12-27 22:41:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/73/627573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/627573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/627573/1"}}},"commit":{"parents":[{"commit":"265e73e62f4f0d0b64825f46449d051a1f23d666","subject":"Merge \"Use renamed template \u0027integrated-gate-py3\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/265e73e62f4f0d0b64825f46449d051a1f23d666"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 22:40:12.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 22:40:12.000000000","tz":60},"subject":"Add puppetforge publish and add to puppet release","message":"Add puppetforge publish and add to puppet release\n\nThis adds a new publish playbook that uses the new\nupload-puppetforge role that is added in [1].\n\n[1] https://review.openstack.org/#/c/627553/\n\nChange-Id: Iec643ec2af8c6c1567653608ef7851a5ac6e414a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f33dae985f9c58bba6bb3f9e39d9e4ec33cc9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f33dae985f9c58bba6bb3f9e39d9e4ec33cc9af"}]},"branch":"refs/heads/master"},"5aa06c2e95f730ad99cd6eb519549261539833cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-12-27 22:41:39.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/73/627573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/627573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/627573/2"}}},"commit":{"parents":[{"commit":"265e73e62f4f0d0b64825f46449d051a1f23d666","subject":"Merge \"Use renamed template \u0027integrated-gate-py3\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/265e73e62f4f0d0b64825f46449d051a1f23d666"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 22:40:12.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 22:41:22.000000000","tz":60},"subject":"Add puppetforge publish and add to puppet release","message":"Add puppetforge publish and add to puppet release\n\nThis adds a new publish playbook that uses the new\nupload-puppetforge role that is added in [1].\n\n[1] https://review.openstack.org/#/c/627553/\n\nDepends-On: https://review.openstack.org/#/c/627553/\nChange-Id: Iec643ec2af8c6c1567653608ef7851a5ac6e414a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5aa06c2e95f730ad99cd6eb519549261539833cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5aa06c2e95f730ad99cd6eb519549261539833cf"}]},"branch":"refs/heads/master"},"eb0347173c1d091b91378bf22c108a0f71118789":{"kind":"REWORK","_number":3,"created":"2019-01-12 15:53:43.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/73/627573/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/627573/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/627573/3"}}},"commit":{"parents":[{"commit":"265e73e62f4f0d0b64825f46449d051a1f23d666","subject":"Merge \"Use renamed template \u0027integrated-gate-py3\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/265e73e62f4f0d0b64825f46449d051a1f23d666"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 22:40:12.000000000","tz":60},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-01-12 15:52:41.000000000","tz":0},"subject":"Add puppetforge publish and add to puppet release","message":"Add puppetforge publish and add to puppet release\n\nThis adds a new publish playbook that uses the new\nupload-puppetforge role that is added in [1].\n\n[1] https://review.openstack.org/#/c/627553/\n\nDepends-On: https://review.openstack.org/#/c/627553/\nChange-Id: Iec643ec2af8c6c1567653608ef7851a5ac6e414a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb0347173c1d091b91378bf22c108a0f71118789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb0347173c1d091b91378bf22c108a0f71118789"}]},"branch":"refs/heads/master"},"e2975095cf547007a4fe27b853b51eb47df10456":{"kind":"REWORK","_number":4,"created":"2019-01-14 08:34:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/73/627573/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/627573/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/627573/4"}}},"commit":{"parents":[{"commit":"265e73e62f4f0d0b64825f46449d051a1f23d666","subject":"Merge \"Use renamed template \u0027integrated-gate-py3\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/265e73e62f4f0d0b64825f46449d051a1f23d666"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 22:40:12.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-01-14 08:34:02.000000000","tz":60},"subject":"Add puppetforge publish and add to puppet release","message":"Add puppetforge publish and add to puppet release\n\nThis adds a new publish playbook that uses the new\nupload-puppetforge role that is added in [1].\n\n[1] https://review.openstack.org/#/c/627553/\n\nDepends-On: https://review.openstack.org/#/c/627553/\nChange-Id: Iec643ec2af8c6c1567653608ef7851a5ac6e414a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2975095cf547007a4fe27b853b51eb47df10456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2975095cf547007a4fe27b853b51eb47df10456"}]},"branch":"refs/heads/master"},"afbc57f93b3b84dab502b1fa85d5dc40294bc08e":{"kind":"REWORK","_number":5,"created":"2019-01-14 11:32:58.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/73/627573/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/627573/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/627573/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/627573/5"}}},"commit":{"parents":[{"commit":"265e73e62f4f0d0b64825f46449d051a1f23d666","subject":"Merge \"Use renamed template \u0027integrated-gate-py3\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/265e73e62f4f0d0b64825f46449d051a1f23d666"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-12-27 22:40:12.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-01-14 11:32:51.000000000","tz":60},"subject":"Add puppetforge publish and add to puppet release","message":"Add puppetforge publish and add to puppet release\n\nThis adds a new publish playbook that uses the new\nupload-puppetforge role that is added in [1].\n\n[1] https://review.openstack.org/#/c/627553/\n\nDepends-On: https://review.openstack.org/#/c/627553/\nChange-Id: Iec643ec2af8c6c1567653608ef7851a5ac6e414a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/afbc57f93b3b84dab502b1fa85d5dc40294bc08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/afbc57f93b3b84dab502b1fa85d5dc40294bc08e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
