)]}'
{"id":"openstack%2Fproject-config~636574","triplet_id":"openstack%2Fproject-config~master~Ia474af67ff89795a561579c332e276bf65f52411","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ia474af67ff89795a561579c332e276bf65f52411","subject":"Update karbor to use storyboard","status":"MERGED","created":"2019-02-13 09:18:32.000000000","updated":"2019-06-11 16:34:55.000000000","submitted":"2019-06-11 16:34:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636574-1560270895619-41fe81b9","meta_rev_id":"55b5b02fb84c05d8d55c0a8693e215dcbb3b2f4a","_number":636574,"virtual_id_number":636574,"owner":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":2,"date":"2019-06-11 16:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 16:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 16:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-18 23:28:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-06-06 19:40:27.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-11 16:09:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-06-11 16:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0519ee4927561c1127d85ad88a0204a27c51e1d8","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-02-13 09:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cc8af5678209595d2c030b5809b95ac7f018930","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 09:25:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/636574/1/check/openstack-zuul-jobs-linters/5174618/ : SUCCESS in 6m 15s\n- project-config-gerrit http://logs.openstack.org/74/636574/1/check/project-config-gerrit/3b0f182/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"273a929df5789c42f4535dfcd7d67b4b604f03be","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-02-13 09:43:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cfb31a1beb3b600dcecc3e2b2af88caa8525c993","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-02-18 23:28:04.000000000","message":"Patch Set 1: Code-Review+2\n\nNot approving as this may need a coordinated cut over.","accounts_in_message":[],"_revision_number":1},{"id":"3b5c144626d62126da93f67a5de2928a2a491355","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-02-19 02:03:00.000000000","message":"Patch Set 1:\n\nIt looks like https://bugs.launchpad.net/karbor and https://bugs.launchpad.net/karbor-dashboard have bugs which should be imported. Please get up with me (fungi) in IRC to work out when you want to have the cutover happen.","accounts_in_message":[],"_revision_number":1},{"id":"c7353a9916d649c0ea3e146cdccb2c9412e7f426","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-08 18:00:39.000000000","message":"Patch Set 1: -Code-Review\n\nJiao, please see Jeremy\u0027s comment and act.","accounts_in_message":[],"_revision_number":1},{"id":"b00fffb92475bd4b436300946c9876ed8c670a00","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-06 19:40:27.000000000","message":"Patch Set 1:\n\nJiao, please see comments and react...","accounts_in_message":[],"_revision_number":1},{"id":"a1de277676caff9a7d469c36bafba8d85f2b20d6","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-06-11 09:32:15.000000000","message":"Patch Set 1:\n\n\u003e Jiao, please see comments and react...\n\nThanks for reminder me, I will contact with fungi on irc quickly.","accounts_in_message":[],"_revision_number":1},{"id":"b64e6ba3e240127b047557d61bcebaf5317a646c","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-06-11 11:55:10.000000000","message":"Patch Set 1:\n\n\u003e It looks like https://bugs.launchpad.net/karbor and\n \u003e https://bugs.launchpad.net/karbor-dashboard have bugs which should\n \u003e be imported. Please get up with me (fungi) in IRC to work out when\n \u003e you want to have the cutover happen.\n\nHi fungi, I don\u0027t know if you get my \u0027ping\u0027 on irc. Bugs on the launchpad are too old now (or some of them are not useful, but I have no right to mark them ), so I think we do not need import them on storyboard. We can start using storyboard now. Let\u0027s start a new day for karbor :) thanks very much","accounts_in_message":[],"_revision_number":1},{"id":"932c60cc2dea58b7428b5e4f661e435b8c515696","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-11 15:44:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis should be fine to merge now in that case. Thanks for following up!\n\nYou may also want to consider adding these three projects to a common karbor project-group (using a \"groups\" list in each project specifying \"karbor\" as the only list entry), but that can be done with a new change any time in the future.","accounts_in_message":[],"_revision_number":1},{"id":"cbccb042830cf89b808f54033d85ac4b7cfb0f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:46:06.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"561390824538b23547d3ce2a1eba969db0db5140","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-11 15:52:00.000000000","message":"Patch Set 1:\n\nAlso, I still strongly encourage you to reach out to the administrators listed at https://launchpad.net/~karbor-drivers/+members and see if one of them can get you added to that group so that you can update the project descriptions on launchpad and close down bug reporting there to avoid future confusion. See https://docs.openstack.org/infra/storyboard/migration.html#recently-migrated for recommendations.","accounts_in_message":[],"_revision_number":1},{"id":"dc9d46aa8c095bebdb14046fc4fe60d9ae57c6ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:52:13.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"97d5350ca7f1a23c65ba764f296036e5078c2d86","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-11 16:02:40.000000000","message":"Patch Set 1: Workflow-1\n\nSince this appears to require a rebase to be able to merge, I\u0027m going to take the initiative to add a \"karbor\" project group for you while I\u0027m doing that.","accounts_in_message":[],"_revision_number":1},{"id":"d74c029e8132c7e8d828f7ab378ce6477e1920ce","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-11 16:08:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1b5d0f0ea3a1a8b3e1fbe53d9db9693f3a4b6a8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-11 16:09:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nReapproving now.","accounts_in_message":[],"_revision_number":2},{"id":"347bdebd7cd5b75bc1a43bccd3f8343887988847","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 16:21:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/636574/2/check/openstack-zuul-jobs-linters/5641c59/ : SUCCESS in 5m 53s\n- project-config-gerrit http://logs.openstack.org/74/636574/2/check/project-config-gerrit/ea81188/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2b53243d2f97573e7260f011cf6bf3645fe32bc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 16:21:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c270095315131aae05739b36cab02b1f943f3a02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 16:34:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"55b5b02fb84c05d8d55c0a8693e215dcbb3b2f4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 16:34:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/636574/2/gate/openstack-zuul-jobs-linters/e6216ff/ : SUCCESS in 8m 12s\n- project-config-gerrit http://logs.openstack.org/74/636574/2/gate/project-config-gerrit/80f6e5c/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e4280c1bcd17dea616ff017e17f97c953b90aee","revisions":{"187cd2df4e49a594b9abeb4738d0342649ec3672":{"kind":"REWORK","_number":1,"created":"2019-02-13 09:18:32.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/74/636574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/636574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/636574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/636574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/636574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/636574/1"}}},"commit":{"parents":[{"commit":"de37bfd6f6a9396b780799d278bc7c7ae1bfd004","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de37bfd6f6a9396b780799d278bc7c7ae1bfd004"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-02-13 09:17:05.000000000","tz":480},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-02-13 09:18:26.000000000","tz":480},"subject":"Update karbor to use storyboard","message":"Update karbor to use storyboard\n\nChange-Id: Ia474af67ff89795a561579c332e276bf65f52411\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/187cd2df4e49a594b9abeb4738d0342649ec3672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/187cd2df4e49a594b9abeb4738d0342649ec3672"}]},"branch":"refs/heads/master"},"2e4280c1bcd17dea616ff017e17f97c953b90aee":{"kind":"REWORK","_number":2,"created":"2019-06-11 16:08:53.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/74/636574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/636574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/636574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/636574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/636574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/636574/2"}}},"commit":{"parents":[{"commit":"0b32e0a8d8a09abc9ddb21d02e9b43740ec1c4f8","subject":"Merge \"Remove glean config for OVH nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b32e0a8d8a09abc9ddb21d02e9b43740ec1c4f8"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-02-13 09:17:05.000000000","tz":480},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-06-11 16:08:37.000000000","tz":0},"subject":"Update karbor to use storyboard","message":"Update karbor to use storyboard\n\nChange-Id: Ia474af67ff89795a561579c332e276bf65f52411\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2e4280c1bcd17dea616ff017e17f97c953b90aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2e4280c1bcd17dea616ff017e17f97c953b90aee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
