)]}'
{"id":"openstack%2Fproject-config~638583","triplet_id":"openstack%2Fproject-config~master~I39c055ec148670e015c9b56a015fd8c95839e044","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I39c055ec148670e015c9b56a015fd8c95839e044","subject":"Update nodepool dib stats","status":"MERGED","created":"2019-02-22 05:14:37.000000000","updated":"2019-12-17 06:10:44.000000000","submitted":"2019-12-17 06:10:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"638583-1576563044487-69f755c7","meta_rev_id":"3b0ccdecdf56858d514302e6906a9532aad62a60","_number":638583,"virtual_id_number":638583,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 06:10:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":2,"date":"2019-12-17 05:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":1,"date":"2019-12-17 05:53:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-22 07:19:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-08 17:19:45.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-09 12:54:35.000000000","updated_by":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"reviewer":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"state":"REVIEWER"},{"updated":"2019-12-17 06:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88a5c7c792bea7fd2b04968e5377852707da8de2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-02-22 05:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fb3f4660c5cafb3018531e040cfde2134bb3bb3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-02-22 05:18:11.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"524cf3fcb8ddf2e3240aa8ffd247d51a560365b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 05:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/83/638583/1/check/openstack-zuul-jobs-linters/cdcbd5c/ : SUCCESS in 6m 20s\n- project-config-grafyaml http://logs.openstack.org/83/638583/1/check/project-config-grafyaml/5b96de0/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"517a9ec8e0f1ce930e31eed9d951530572c2b624","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-02-22 07:19:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8bd0be271f41b1d5af26054b6aaf71f1570ef59a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-08 17:19:45.000000000","message":"Patch Set 1: Code-Review+2\n\nNote the depends on has merged but we haven\u0027t redeployed zuul to use that commit yet.","accounts_in_message":[],"_revision_number":1},{"id":"e1408e932d122c5eee9a04ea2e1191cca0488af7","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2019-04-09 12:54:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"827fa3b7cc3432ca02381a2afb7880cbd0f133b7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-28 18:14:54.000000000","message":"Patch Set 1:\n\ndo we run now zuul with these changes and can merge this?","accounts_in_message":[],"_revision_number":1},{"id":"7cbae48675e74b5d64e67bf9498d696859ce8e77","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-12-17 05:52:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a45fc5e92d5cedd16e51bbbfc504d85d21d241ea","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-12-17 05:53:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\ntaking prior reviews, these stats are live","accounts_in_message":[],"_revision_number":2},{"id":"1763a8c20170f4f3801f4e4f06c93bf702969add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 06:01:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bf6f960ddce54c8c83e398f0433609a3 : SUCCESS in 7m 57s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/7aa7af2a5ef24095bbc44ec52c69b9a4 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"993807816fccf7e6983d8740969c87cb16f074ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 06:02:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"124884f6d41782b0e532928d4c8812086ba0d223","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 06:10:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3b0ccdecdf56858d514302e6906a9532aad62a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 06:10:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d01a7c6ae3b7447fb640042677d06161 : SUCCESS in 7m 25s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/15bbfb9ab2f548f792c7debe958469e5 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8938e5e250e90e5f34e13d81fe9848999e073de","revisions":{"e8b09498ea54b2684260cb5b3c3184fb14b8192c":{"kind":"REWORK","_number":1,"created":"2019-02-22 05:14:37.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/83/638583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/638583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/638583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/638583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/638583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/638583/1"}}},"commit":{"parents":[{"commit":"c8582e4d26237b26076ec9598e55b88d22bb2b58","subject":"Add nodepool-dib dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8582e4d26237b26076ec9598e55b88d22bb2b58"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-02-22 05:13:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-02-22 05:13:14.000000000","tz":660},"subject":"Update nodepool dib stats","message":"Update nodepool dib stats\n\nThis updates the stats for the changes made in the dependent change.\n\nChange-Id: I39c055ec148670e015c9b56a015fd8c95839e044\nDepends-On: https://review.openstack.org/638533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8b09498ea54b2684260cb5b3c3184fb14b8192c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8b09498ea54b2684260cb5b3c3184fb14b8192c"}]},"branch":"refs/heads/master"},"e8938e5e250e90e5f34e13d81fe9848999e073de":{"kind":"REWORK","_number":2,"created":"2019-12-17 05:52:39.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/83/638583/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/638583/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/638583/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/638583/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/638583/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/638583/2"}}},"commit":{"parents":[{"commit":"71a740c354fc060c3ede6b1a863f2381959b1b27","subject":"Update Neutron\u0027s Grafana dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/71a740c354fc060c3ede6b1a863f2381959b1b27"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-02-22 05:13:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-12-17 05:52:24.000000000","tz":660},"subject":"Update nodepool dib stats","message":"Update nodepool dib stats\n\nThis updates the stats for the changes made in the dependent change.\n\nChange-Id: I39c055ec148670e015c9b56a015fd8c95839e044\nDepends-On: https://review.openstack.org/638533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8938e5e250e90e5f34e13d81fe9848999e073de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8938e5e250e90e5f34e13d81fe9848999e073de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
