)]}'
{"id":"openstack%2Fproject-config~645574","triplet_id":"openstack%2Fproject-config~master~Ia55d240fa8ce1d19ef69608e68c47e02776f15ca","project":"openstack/project-config","branch":"master","topic":"add-buster-image","hashtags":[],"change_id":"Ia55d240fa8ce1d19ef69608e68c47e02776f15ca","subject":"Add a Debian Buster image.","status":"MERGED","created":"2019-03-22 13:36:18.000000000","updated":"2019-05-21 22:30:54.000000000","submitted":"2019-05-21 22:30:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"645574-1558477854956-03399c29","meta_rev_id":"db4c0b8a839023970b7a2e1819cdcad0b5c26566","_number":645574,"virtual_id_number":645574,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-05-21 22:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 22:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2019-05-21 22:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2019-05-21 22:13:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 22:00:28.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-05-21 22:13:07.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-05-21 22:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f57c6c3c91476cba6f6ec0ce6d7910390f5aa829","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-03-22 13:36:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff6b340fb5e7f382b2f7ca0f5b72dc0ee9fe3ce8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 13:46:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/645574/1/check/openstack-zuul-jobs-linters/3e5413a/ : SUCCESS in 7m 07s\n- project-config-grafyaml http://logs.openstack.org/74/645574/1/check/project-config-grafyaml/0b921ea/ : SUCCESS in 4m 25s\n- project-config-nodepool http://logs.openstack.org/74/645574/1/check/project-config-nodepool/aa8273c/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f240093a4f2ee2010c79da1decdfc6c260c18364","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-03-22 13:49:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e807813d579e363c5686e1459d5a31e223f83a7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-03-24 23:27:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI\u0027m not opposed to this in any way ... but we don\u0027t have the mirror setup, and I don\u0027t think that DIB has any testing of this in it\u0027s gate at all either, so I couldn\u0027t confidently say it works.  we should add it to the debian nodepool job to test at least basic boot compatibility before adding to production.","accounts_in_message":[],"_revision_number":1},{"id":"21a3b7483bfc915ed1a6020b199eb0960979100a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-03-25 00:06:56.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f2fb7a0b06814e13828bb788df8df8a1517343a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-03-29 13:04:29.000000000","message":"Patch Set 1: Code-Review+2\n\nRestoring my +2 now that we have http://files.openstack.org/mirror/debian/dists/buster/ in place.","accounts_in_message":[],"_revision_number":1},{"id":"8d70066011f8dd21cd58bc6ef53c25291d29fbc1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-04-02 05:16:20.000000000","message":"Patch Set 1:\n\nSo it looks like buster isn\u0027t yet quite ready to build [1] ...\n\nI\u0027ve not seen this particular error from debootstrap before:\n\n 2019-04-02 04:47:36.934 | W: Failure while installing base packages.  This will be re-attempted up to five times.\n 2019-04-02 04:47:36.938 | W: See /opt/dib/tmp/dib_build.C5uy1wal/mnt/debootstrap/debootstrap.log for details\n\nwe don\u0027t actually capture that logfile; this will take more investigation to see what\u0027s actually happening.\n\n[1] http://logs.openstack.org/14/649214/1/experimental/nodepool-functional-py35-debian-src/3d7f36a/controller/logs/builds/debian-buster-0000000002_log.txt.gz#_2019-04-02_04_47_36_934","accounts_in_message":[],"_revision_number":1},{"id":"72a101fc37441ec5beb3952e7be4a2b4e5ccd0a9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-04-03 03:18:03.000000000","message":"Patch Set 1:\n\nI have investigated this enough to understand what\u0027s going on, and have opened [1].  With [2] dib can seem to build the images; however the question is how to get the supported debootstrap.  Either we have to get it merged via upstream, or create something like a local ppa and the arrange for it to be installed on CI nodes and all the builders.\n\n[1] https://bugs.launchpad.net/diskimage-builder/+bug/1822927\n[2] https://review.openstack.org/649496","accounts_in_message":[],"_revision_number":1},{"id":"2788fa312eee457212238cc84fec6ae4b580c35e","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-05-21 11:47:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"230cb0372f613241f52eb826c196c76e8e75a23d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 12:01:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/645574/2/check/openstack-zuul-jobs-linters/23e069d/ : SUCCESS in 6m 24s\n- project-config-grafyaml http://logs.openstack.org/74/645574/2/check/project-config-grafyaml/fae0adb/ : SUCCESS in 3m 09s\n- project-config-nodepool http://logs.openstack.org/74/645574/2/check/project-config-nodepool/b6f4ae9/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"cc7bda6258ba458bc0f38f9ce66d45fd4f090b2a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-21 22:00:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"14984fdd1726151704bf06c5ef1799aa1f338b15","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-21 22:13:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a11012a408b344c6e6055a7cbbd20bcc0af232e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 22:13:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"986901441264dce97c2195c7879f548ea9b083de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 22:30:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/74/645574/2/gate/openstack-zuul-jobs-linters/dae350e/ : SUCCESS in 6m 48s\n- project-config-grafyaml http://logs.openstack.org/74/645574/2/gate/project-config-grafyaml/fdaae19/ : SUCCESS in 3m 06s\n- project-config-nodepool http://logs.openstack.org/74/645574/2/gate/project-config-nodepool/61c3a9f/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ad11b996262bab5740b7b56a4ccea4e2c818e91f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 22:30:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bd7b5f0445ddd237dbb5ff4f7c28505ac36ad5e","revisions":{"ca8d08146f0406a9011469e490c3b421935f6fb6":{"kind":"REWORK","_number":1,"created":"2019-03-22 13:36:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/74/645574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/645574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/645574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/645574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/645574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/645574/1"}}},"commit":{"parents":[{"commit":"90f1dd4975bdb1f7b46680a1b9e2bfb1c9ce13d6","subject":"Cleanup packethost zk data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90f1dd4975bdb1f7b46680a1b9e2bfb1c9ce13d6"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-03-22 13:35:55.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-03-22 13:35:55.000000000","tz":60},"subject":"Add a Debian Buster image.","message":"Add a Debian Buster image.\n\nChange-Id: Ia55d240fa8ce1d19ef69608e68c47e02776f15ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca8d08146f0406a9011469e490c3b421935f6fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca8d08146f0406a9011469e490c3b421935f6fb6"}]},"branch":"refs/heads/master"},"7bd7b5f0445ddd237dbb5ff4f7c28505ac36ad5e":{"kind":"REWORK","_number":2,"created":"2019-05-21 11:47:07.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/74/645574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/645574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/645574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/645574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/645574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/645574/2"}}},"commit":{"parents":[{"commit":"c690d65bdbc2c18adb13d5d018cc74daa5ae75de","subject":"Merge \"Remove unused job publish-openstack-sphinx-docs-direct\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c690d65bdbc2c18adb13d5d018cc74daa5ae75de"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-03-22 13:35:55.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-05-21 11:47:04.000000000","tz":120},"subject":"Add a Debian Buster image.","message":"Add a Debian Buster image.\n\nChange-Id: Ia55d240fa8ce1d19ef69608e68c47e02776f15ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bd7b5f0445ddd237dbb5ff4f7c28505ac36ad5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bd7b5f0445ddd237dbb5ff4f7c28505ac36ad5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
