)]}'
{"id":"openstack%2Fproject-config~649082","triplet_id":"openstack%2Fproject-config~master~I91e09035ff829392306fd970aad439fdb18ae1a8","project":"openstack/project-config","branch":"master","topic":"retire-cloud-init","hashtags":[],"change_id":"I91e09035ff829392306fd970aad439fdb18ae1a8","subject":"Retire cloud-init","status":"MERGED","created":"2019-04-01 15:51:03.000000000","updated":"2019-05-06 07:26:16.000000000","submitted":"2019-05-06 07:26:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649082-1557127576877-ba8cb0da","meta_rev_id":"9d3b75773edc36ae930e858d53969c8db237a563","_number":649082,"virtual_id_number":649082,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2019-05-02 19:50:18.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-06 07:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"all":[{"value":1,"date":"2019-05-02 17:09:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-06 07:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-06 06:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-05-06 07:15:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-02 11:49:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-04-05 19:55:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-14 21:50:31.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-02 17:09:10.000000000","updated_by":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"reviewer":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"state":"REVIEWER"},{"updated":"2019-05-06 06:30:53.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-05-06 07:15:13.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-06 07:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dabb0561da24ffe555c0d6431b57640f2400c6c1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-01 15:51:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1832ccde34853b52144cb07e0af3a240aaa93224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 16:03:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/82/649082/1/check/openstack-zuul-jobs-linters/768226a/ : SUCCESS in 7m 17s\n- project-config-gerrit http://logs.openstack.org/82/649082/1/check/project-config-gerrit/283d9f5/ : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"44c0759874cb31cefbb2d5629c99aab52e81a7c5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-02 11:10:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"425e630cd6e28d479b69376ce6017be5f2721d3d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-04-02 11:49:36.000000000","message":"Patch Set 1:\n\nFYI, this is a duplicate of https://review.openstack.org/392903 but I think it makes sense to abandon that patch and push this one through now.","accounts_in_message":[],"_revision_number":1},{"id":"38af40b66d772cd9da64929bdf265f07a20c239a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-02 13:54:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"340ecbe77344bf535b3c0d68e9f05c495143bb8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 14:13:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/82/649082/2/check/openstack-zuul-jobs-linters/ab17159/ : SUCCESS in 7m 30s\n- project-config-gerrit http://logs.openstack.org/82/649082/2/check/project-config-gerrit/2b8052c/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"36c603cc2b36308fd190a892802337f5af2f5082","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-05 19:55:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f596718817deb64bdce3673dba641b1511099a25","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-14 21:50:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d523fc8230e0217578232de2778c1e72b0920cb6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-30 20:58:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b7f56e59169df0a558737fb094a94a6a1ba5069","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-30 20:59:08.000000000","message":"Patch Set 2:\n\nHad to rebase post-move","accounts_in_message":[],"_revision_number":2},{"id":"13c52cb90fcac4fe56b7713a5954af49e3d328b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 21:16:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/82/649082/3/check/openstack-zuul-jobs-linters/fe94c37/ : SUCCESS in 6m 33s\n- project-config-gerrit http://logs.openstack.org/82/649082/3/check/project-config-gerrit/dabd1b6/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"20e1d5b51059c78f52d5ac57247cd612906d2721","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2019-05-02 17:09:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"75476e1ad393170d568a8a36b0cc50d0ec2136d7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-02 19:50:18.000000000","message":"Patch Set 3:\n\nCan someone take a look at this and abandon [1]? I assume someone also needs to close all the open reviews on Gerrit against this project, or does that happen automatically?\n\n[1] https://review.opendev.org/#/c/392903/","accounts_in_message":[],"_revision_number":3},{"id":"c08bd48199c2437a37badaebe62e3033e2a70520","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-03 07:04:18.000000000","message":"Patch Set 3:\n\nYes, somebody needs to abandon those open reviews - best ask on #openstack-infra","accounts_in_message":[],"_revision_number":3},{"id":"d72e8c9a39b02d8528535ed321c6e24a29a13843","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-03 07:05:09.000000000","message":"Patch Set 3:\n\nJoshua, can you abandon these https://review.opendev.org/#/q/project:x/cloud-init+is:open , please?","accounts_in_message":[],"_revision_number":3},{"id":"14c28430cf05cf8fd45629d4d42ea0526f1cfe5d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-05-06 06:30:53.000000000","message":"Patch Set 3: Code-Review+2\n\nAbandoned the open changes with a note from extant readme and a link to this.  LGTM","accounts_in_message":[],"_revision_number":3},{"id":"468499eb7b2476f6fcfb034443521c31dc7897aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-06 07:15:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9779fe1df9a715143af1bf6a6fd88b2034b45f18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 07:15:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8b11af00fc9c9a2a148c95ae7f049d6aab3c8a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 07:26:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/82/649082/3/gate/openstack-zuul-jobs-linters/1b42c4c/ : SUCCESS in 6m 00s\n- project-config-gerrit http://logs.openstack.org/82/649082/3/gate/project-config-gerrit/0a64541/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9b5ce010b385c25ae4b6ea71067209db4e5c2253","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 07:26:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b6886279d2b8e4899b99b6b47eff9d5d0bef7736","revisions":{"96bb38a4f8bc2573261bcbe466a7d307b738b351":{"kind":"REWORK","_number":1,"created":"2019-04-01 15:51:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/649082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/649082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/649082/1"}}},"commit":{"parents":[{"commit":"bcb46f4999d6cb7daedf830598db23fd2ec9ec76","subject":"Merge \"Trove: use storyboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bcb46f4999d6cb7daedf830598db23fd2ec9ec76"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-01 15:49:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-01 15:50:40.000000000","tz":60},"subject":"Retire cloud-init","message":"Retire cloud-init\n\nThis was never an official project so no patches against governance are\nrequired.\n\nChange-Id: I91e09035ff829392306fd970aad439fdb18ae1a8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: https://review.openstack.org/649081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96bb38a4f8bc2573261bcbe466a7d307b738b351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96bb38a4f8bc2573261bcbe466a7d307b738b351"}]},"branch":"refs/heads/master"},"960a9272073795630816d03eb5230177c8c96069":{"kind":"REWORK","_number":2,"created":"2019-04-02 13:54:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/649082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/649082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/649082/2"}}},"commit":{"parents":[{"commit":"bcb46f4999d6cb7daedf830598db23fd2ec9ec76","subject":"Merge \"Trove: use storyboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bcb46f4999d6cb7daedf830598db23fd2ec9ec76"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-01 15:49:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-02 13:54:39.000000000","tz":60},"subject":"Retire cloud-init","message":"Retire cloud-init\n\nThis was never an official project so no patches against governance are\nrequired.\n\nChange-Id: I91e09035ff829392306fd970aad439fdb18ae1a8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: https://review.openstack.org/649081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/960a9272073795630816d03eb5230177c8c96069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/960a9272073795630816d03eb5230177c8c96069"}]},"branch":"refs/heads/master"},"b6886279d2b8e4899b99b6b47eff9d5d0bef7736":{"kind":"REWORK","_number":3,"created":"2019-04-30 20:58:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/649082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/649082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/649082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/649082/3"}}},"commit":{"parents":[{"commit":"139c38acdc088eb9d3705d1440ef6b5280319936","subject":"Merge \"Stop master translations for tripleo-ui\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/139c38acdc088eb9d3705d1440ef6b5280319936"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-01 15:49:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-30 20:58:37.000000000","tz":-360},"subject":"Retire cloud-init","message":"Retire cloud-init\n\nThis was never an official project so no patches against governance are\nrequired.\n\nChange-Id: I91e09035ff829392306fd970aad439fdb18ae1a8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: https://review.openstack.org/649081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6886279d2b8e4899b99b6b47eff9d5d0bef7736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6886279d2b8e4899b99b6b47eff9d5d0bef7736"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
