)]}'
{"id":"openstack%2Fproject-config~650422","triplet_id":"openstack%2Fproject-config~master~Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08","project":"openstack/project-config","branch":"master","topic":"retire-osa-unused-roles","hashtags":[],"change_id":"Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08","subject":"Retire openstack-ansible unused roles","status":"MERGED","created":"2019-04-05 18:49:09.000000000","updated":"2019-06-24 23:20:22.000000000","submitted":"2019-06-24 23:20:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650422-1561418422508-f9ed76da","meta_rev_id":"9af93e838ece50f6c849c079190384fc98027139","_number":650422,"virtual_id_number":650422,"owner":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-06-24 23:20:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-06-24 22:52:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-06-23 19:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-06-24 22:52:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-23 19:29:01.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-24 22:52:14.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-06-24 23:20:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b63e3c0de72859dcbea55e5e6e51d09346ce2ff5","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-05 18:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35b56d6464492d369145f2640dd6e055194415fc","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-05 18:53:32.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a597cb18ff9156a87e5015d380b9ad03945be8f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 19:02:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/22/650422/2/check/openstack-zuul-jobs-linters/ed411ed/ : SUCCESS in 7m 24s\n- project-config-gerrit http://logs.openstack.org/22/650422/2/check/project-config-gerrit/7ded246/ : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6845212f5454af40eb0c8c7738a3500d757d82d5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-05 20:26:59.000000000","message":"Patch Set 2: Code-Review-1\n\nI believe you have skipped the noop job step and content retirement as described at https://docs.openstack.org/infra/manual/drivers.html#step-2-end-project-gating.","accounts_in_message":[],"_revision_number":2},{"id":"9f0a13bd5b500380d5be45d5309fd7fc6cf0426a","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-05 20:49:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"34bf307b5f70cd5a43270a78e759e1d53c780a11","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-05 20:51:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ae9ddcee97d40993c68e87c47abbb4f05f426c6b","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-05 20:55:15.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0a4cedc196860a0918dca4cf5f32d985b061ad88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 21:07:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/22/650422/5/check/openstack-zuul-jobs-linters/b902219/ : SUCCESS in 6m 45s\n- project-config-gerrit http://logs.openstack.org/22/650422/5/check/project-config-gerrit/96b6378/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":5},{"id":"83a184de84e0e6a1dd71ae5e18546075ec0229a9","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-05 23:59:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9a29538d19cd6fa2f27f8b58df3806ac864e66d7","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-06 00:00:10.000000000","message":"Patch Set 6: Workflow-1\n\nthis is step 4 as stated here https://docs.openstack.org/infra/manual/drivers.html#step-4-remove-project-from-infrastructure-systems\n\nwait to be merged https://review.openstack.org/#/c/650477/","accounts_in_message":[],"_revision_number":6},{"id":"b110545e496817afd4feac4f8893b3173c84cca9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-06 00:12:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/22/650422/6/check/openstack-zuul-jobs-linters/2802e89/ : SUCCESS in 7m 25s\n- project-config-gerrit http://logs.openstack.org/22/650422/6/check/project-config-gerrit/8d49a7c/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":6},{"id":"6995a3189fea0a76dd3b8d1429795f4e6f4efc88","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-18 00:02:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"31dd4da28b4f65119ddde818a3670c049e004090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 00:10:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/22/650422/7/check/openstack-zuul-jobs-linters/8920725/ : SUCCESS in 6m 44s\n- project-config-gerrit http://logs.openstack.org/22/650422/7/check/project-config-gerrit/2085582/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":7},{"id":"72544881c01b7da0ab9968e21f32f45a3bce6b22","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-18 06:11:31.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)\n\nYou first need to retire the repos, let\u0027s mark the change as WIP until this is done.","accounts_in_message":[],"_revision_number":7},{"id":"cf31e3317198bd803bd91e58d3e44490953212dd","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-18 13:40:30.000000000","message":"Patch Set 7:\n\nThanks Andreas! Asking to the osa team to review the patches that removed the contents","accounts_in_message":[],"_revision_number":7},{"id":"5201dfc9c017285747a3376cb1805be9f0d716a2","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-18 13:40:56.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"c7f4fd9edbb2cb25785a5a1de7ce4cfcc306ff0a","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-18 13:41:08.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"597b82a54ba507fe014ee351ec76b5412a487f68","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-18 13:45:04.000000000","message":"Patch Set 8:\n\nWhy did you add the [WIP]? That is really not needed, Workflow -1 is enough. Now you pushed a change that needs automatic review, and then anothre one to remove it.\n\nYou could add \"Depends-On: \" to the changes that need merging first, see https://docs.openstack.org/infra/manual/developers.html#cross-repository-dependencies","accounts_in_message":[],"_revision_number":8},{"id":"9b64fd97a6880484928d8c3e134a081880a4d8d9","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-18 13:59:13.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"6fd8d4347ddc9a1bb89bfdaa98142818e81e27d5","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-18 13:59:23.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"e4b316321033543f9b40e63a82506ab1ffece6f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 14:14:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/22/650422/9/check/openstack-zuul-jobs-linters/cb05a2e/ : SUCCESS in 8m 15s\n- project-config-gerrit http://logs.openstack.org/22/650422/9/check/project-config-gerrit/789dd9d/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":9},{"id":"80199947657f35afd79d25f5b21ad20c5b7f0d4d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-01 07:21:43.000000000","message":"Patch Set 9:\n\nWhat is the state here? Could you move this forward, please?","accounts_in_message":[],"_revision_number":9},{"id":"235a3212e14b29a6fcbd18b1e2e94ca883fe3c3d","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-11 20:28:07.000000000","message":"Patch Set 9:\n\n\u003e What is the state here? Could you move this forward, please?\n\nsorry for the delay. Rebasing remaining patches...","accounts_in_message":[],"_revision_number":9},{"id":"9cb60717c52d57f169aef427af5276ee82512756","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-11 21:08:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c26b2098946d19909a2b09af874faa572e8bf957","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:29:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/22/650422/10/check/openstack-zuul-jobs-linters/15d39c1/ : SUCCESS in 5m 53s\n- project-config-gerrit http://logs.openstack.org/22/650422/10/check/project-config-gerrit/3f20c76/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":10},{"id":"65cced54ede6d151517b943ecbfb21712097df8b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-23 19:29:01.000000000","message":"Patch Set 10: Code-Review+2\n\nDependencies are all merged","accounts_in_message":[],"_revision_number":10},{"id":"6ab9771d214821bb70fa93b38c838895cab7833c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-06-24 22:52:14.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"e64a8057afa8db05f5a74ccf0eff1b01ee6bd123","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:52:28.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"f811a26463debbf1fa695a4e6b7ffd14bdf45d5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 23:20:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"fdaf73844bfc1c567e90abb1e2aad14965764331","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 23:20:22.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/22/650422/10/gate/openstack-zuul-jobs-linters/cf7605b/ : SUCCESS in 6m 27s\n- project-config-gerrit http://logs.openstack.org/22/650422/10/gate/project-config-gerrit/907e19c/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"ca1d2e7a89332abbd4d59251e7e066e0491e4510","revisions":{"a2a04ce2f0a55247a7680a73d3c815afeacdf293":{"kind":"REWORK","_number":1,"created":"2019-04-05 18:49:09.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/1"}}},"commit":{"parents":[{"commit":"ca92e62a735e02327879a9ee00b968c94093c486","subject":"Merge \"Retire linaro-cn1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca92e62a735e02327879a9ee00b968c94093c486"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:49:08.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2a04ce2f0a55247a7680a73d3c815afeacdf293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2a04ce2f0a55247a7680a73d3c815afeacdf293"}]},"branch":"refs/heads/master"},"3873941e38ec4d93ccde62300728781c771dd547":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-05 18:53:32.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/2"}}},"commit":{"parents":[{"commit":"ca92e62a735e02327879a9ee00b968c94093c486","subject":"Merge \"Retire linaro-cn1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca92e62a735e02327879a9ee00b968c94093c486"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme  Steinmuller Pimentel","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:53:29.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nDepends-On: https://review.openstack.org/650413\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3873941e38ec4d93ccde62300728781c771dd547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3873941e38ec4d93ccde62300728781c771dd547"}]},"branch":"refs/heads/master"},"f8dea63b913552c0bc8ef44caaa272431e00e770":{"kind":"REWORK","_number":3,"created":"2019-04-05 20:49:17.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/3"}}},"commit":{"parents":[{"commit":"ca92e62a735e02327879a9ee00b968c94093c486","subject":"Merge \"Retire linaro-cn1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca92e62a735e02327879a9ee00b968c94093c486"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 20:49:03.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f8dea63b913552c0bc8ef44caaa272431e00e770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f8dea63b913552c0bc8ef44caaa272431e00e770"}]},"branch":"refs/heads/master"},"3dc40f0502d19629dbff92ea4072040e49db1448":{"kind":"REWORK","_number":4,"created":"2019-04-05 20:51:23.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/4"}}},"commit":{"parents":[{"commit":"ca92e62a735e02327879a9ee00b968c94093c486","subject":"Merge \"Retire linaro-cn1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca92e62a735e02327879a9ee00b968c94093c486"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 20:51:10.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dc40f0502d19629dbff92ea4072040e49db1448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dc40f0502d19629dbff92ea4072040e49db1448"}]},"branch":"refs/heads/master"},"c68ffe3011ea8d436b6a4db7a2374daa6440da8a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-04-05 20:55:15.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/5"}}},"commit":{"parents":[{"commit":"ca92e62a735e02327879a9ee00b968c94093c486","subject":"Merge \"Retire linaro-cn1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca92e62a735e02327879a9ee00b968c94093c486"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 20:53:54.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe patch endings project gate for unsued\nOpenStack Ansible roles for the purposes of\nretiring these roles.\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c68ffe3011ea8d436b6a4db7a2374daa6440da8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c68ffe3011ea8d436b6a4db7a2374daa6440da8a"}]},"branch":"refs/heads/master"},"ca7ffb27bcfc7d7edff426008bac3f67f0668fcd":{"kind":"REWORK","_number":6,"created":"2019-04-05 23:59:16.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/6"}}},"commit":{"parents":[{"commit":"ca92e62a735e02327879a9ee00b968c94093c486","subject":"Merge \"Retire linaro-cn1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca92e62a735e02327879a9ee00b968c94093c486"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 23:59:15.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe patch endings project gate for unsued\nOpenStack Ansible roles for the purposes of\nretiring these roles.\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca7ffb27bcfc7d7edff426008bac3f67f0668fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca7ffb27bcfc7d7edff426008bac3f67f0668fcd"}]},"branch":"refs/heads/master"},"f2d901d758f442ff776fb0b5b860ad2e8560895e":{"kind":"REWORK","_number":7,"created":"2019-04-18 00:02:35.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/7"}}},"commit":{"parents":[{"commit":"47d53f68a83e931f42abed329c6a23f5192638b7","subject":"Temporarily disable inap-mtl01 for maintenance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/47d53f68a83e931f42abed329c6a23f5192638b7"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-18 00:02:34.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe patch endings project gate for unsued\nOpenStack Ansible roles for the purposes of\nretiring these roles.\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2d901d758f442ff776fb0b5b860ad2e8560895e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2d901d758f442ff776fb0b5b860ad2e8560895e"}]},"branch":"refs/heads/master"},"a87bf7396cdff826da65f41f8eb8bef3a1130006":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-04-18 13:40:56.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/8"}}},"commit":{"parents":[{"commit":"47d53f68a83e931f42abed329c6a23f5192638b7","subject":"Temporarily disable inap-mtl01 for maintenance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/47d53f68a83e931f42abed329c6a23f5192638b7"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme  Steinmuller Pimentel","email":"gsteinmuller@vexxhost.com","date":"2019-04-18 13:40:52.000000000","tz":0},"subject":"[WIP] Retire openstack-ansible unused roles","message":"[WIP] Retire openstack-ansible unused roles\n\nThe patch endings project gate for unsued\nOpenStack Ansible roles for the purposes of\nretiring these roles.\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a87bf7396cdff826da65f41f8eb8bef3a1130006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a87bf7396cdff826da65f41f8eb8bef3a1130006"}]},"branch":"refs/heads/master"},"e12d94e22a45a0ce0037e79cb099ff05eaceaa81":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-04-18 13:59:13.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/9"}}},"commit":{"parents":[{"commit":"47d53f68a83e931f42abed329c6a23f5192638b7","subject":"Temporarily disable inap-mtl01 for maintenance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/47d53f68a83e931f42abed329c6a23f5192638b7"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme  Steinmuller Pimentel","email":"gsteinmuller@vexxhost.com","date":"2019-04-18 13:59:08.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe patch endings project gate for unsued\nOpenStack Ansible roles for the purposes of\nretiring these roles.\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nDepends-On: https://review.openstack.org/653631\nDepends-On: https://review.openstack.org/653635\nDepends-On: https://review.openstack.org/653636\nDepends-On: https://review.openstack.org/653621\nDepends-On: https://review.openstack.org/653632\nDepends-On: https://review.openstack.org/653633\nDepends-On: https://review.openstack.org/653630\nNeeded-By: https://review.openstack.org/653643\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e12d94e22a45a0ce0037e79cb099ff05eaceaa81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e12d94e22a45a0ce0037e79cb099ff05eaceaa81"}]},"branch":"refs/heads/master"},"ca1d2e7a89332abbd4d59251e7e066e0491e4510":{"kind":"REWORK","_number":10,"created":"2019-06-11 21:08:16.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/22/650422/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/650422/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/650422/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/650422/10"}}},"commit":{"parents":[{"commit":"2e4280c1bcd17dea616ff017e17f97c953b90aee","subject":"Update karbor to use storyboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2e4280c1bcd17dea616ff017e17f97c953b90aee"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-04-05 18:48:20.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-06-11 21:08:15.000000000","tz":0},"subject":"Retire openstack-ansible unused roles","message":"Retire openstack-ansible unused roles\n\nThe patch endings project gate for unsued\nOpenStack Ansible roles for the purposes of\nretiring these roles.\n\nThe announcement has been made on ML [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004610.html\n\nDepends-On: https://review.openstack.org/653631\nDepends-On: https://review.openstack.org/653635\nDepends-On: https://review.openstack.org/653636\nDepends-On: https://review.openstack.org/653621\nDepends-On: https://review.openstack.org/653632\nDepends-On: https://review.openstack.org/653633\nDepends-On: https://review.openstack.org/653630\nNeeded-By: https://review.openstack.org/653643\n\nChange-Id: Ia71bd9050786c6dc30c494cf0a03e8f3bc5a5f08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca1d2e7a89332abbd4d59251e7e066e0491e4510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca1d2e7a89332abbd4d59251e7e066e0491e4510"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
