)]}'
{"id":"openstack%2Fproject-config~652993","triplet_id":"openstack%2Fproject-config~master~I01e959176dd3b334cb2310da9df10a5ca3fc2315","project":"openstack/project-config","branch":"master","topic":"sdk-job-to-dashboard","hashtags":[],"change_id":"I01e959176dd3b334cb2310da9df10a5ca3fc2315","subject":"Add openstacksdk-functional-devstack-networking job to Neutron dashboard","status":"MERGED","created":"2019-04-16 13:08:56.000000000","updated":"2019-04-24 18:49:21.000000000","submitted":"2019-04-24 18:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652993-1556131761344-332a2ac9","meta_rev_id":"4295be78997f2ae5966d47df1f7e2f4ad22ae327","_number":652993,"virtual_id_number":652993,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 18:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-04-23 12:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-04-24 18:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-04-24 18:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 12:38:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-24 18:31:44.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 18:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9cc03d1c9436da578964f4807a178de781f3656","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-16 13:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b564978fc033fe4220c6e152ca1a16d5406287f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 13:57:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/93/652993/1/check/openstack-zuul-jobs-linters/b844cfb/ : SUCCESS in 6m 41s\n- project-config-grafyaml http://logs.openstack.org/93/652993/1/check/project-config-grafyaml/909bb39/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bf2bdf87fbaca7a3bdabbbc447263205c8583e72","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-16 15:33:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c0029942300fac5ecc072fb1d18d4ec6401ebb8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-21 09:30:55.000000000","message":"Patch Set 1: Code-Review-1\n\nthis needs updating after the OpenDev change, see https://review.opendev.org/654026 - and base on top of it.","accounts_in_message":[],"_revision_number":1},{"id":"81394e0bce5cb55f48b51094e55be6dd5526f6b8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-23 12:31:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45dae922f1becc7d158c5304317f3104d07e594a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-23 12:32:07.000000000","message":"Patch Set 1:\n\n\u003e this needs updating after the OpenDev change, see https://review.opendev.org/654026\n \u003e - and base on top of it.\n\nDone","accounts_in_message":[],"_revision_number":1},{"id":"6da0f5a6a97e6e9016f1c8baee16eb9aee59f78d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-23 12:38:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dd82d9be2562a2e0c6b0f734bb04fe18fa1f310e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:18:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/93/652993/2/check/openstack-zuul-jobs-linters/cb30d18/ : SUCCESS in 7m 30s\n- project-config-grafyaml http://logs.openstack.org/93/652993/2/check/project-config-grafyaml/4379911/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"39b911fd922d4b71bb82bd0aa0576df571c360a7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:31:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"05cdba4f6ac28262cf90a6111da7fb48ebc85686","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:32:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1765b8d603e50e5a5f43f97d676701dd79fcd04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:49:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/93/652993/2/gate/openstack-zuul-jobs-linters/ab6b21d/ : SUCCESS in 7m 49s\n- project-config-grafyaml http://logs.openstack.org/93/652993/2/gate/project-config-grafyaml/a120eae/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6e7ffe517ffecbf43b4512a8147049a2ccb7440b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:49:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"091ecfc3ca049607ca66934c1d7b40ed24e26746","revisions":{"7f2f5e857e92e5ba4608d8d74b296bce4af594b3":{"kind":"REWORK","_number":1,"created":"2019-04-16 13:08:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/652993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/652993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/652993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/652993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/652993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/652993/1"}}},"commit":{"parents":[{"commit":"a5c9ca7761591342790f072996114b5649c367d3","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a5c9ca7761591342790f072996114b5649c367d3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-16 13:08:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-16 13:08:03.000000000","tz":120},"subject":"Add openstacksdk-functional-devstack-networking job to Neutron dashboard","message":"Add openstacksdk-functional-devstack-networking job to Neutron dashboard\n\nThis job was added some time ago to neutron check queue so we should\nhave it in grafana dashboard to be aware if something wrong will happen\nwith it.\n\nChange-Id: I01e959176dd3b334cb2310da9df10a5ca3fc2315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f2f5e857e92e5ba4608d8d74b296bce4af594b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f2f5e857e92e5ba4608d8d74b296bce4af594b3"}]},"branch":"refs/heads/master"},"091ecfc3ca049607ca66934c1d7b40ed24e26746":{"kind":"REWORK","_number":2,"created":"2019-04-23 12:31:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/652993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/652993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/652993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/652993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/652993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/652993/2"}}},"commit":{"parents":[{"commit":"ca1cf6ff02a881da186e6f48ec84da4a9644810f","subject":"Update Airship projects for gerritbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca1cf6ff02a881da186e6f48ec84da4a9644810f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-16 13:08:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-23 12:31:32.000000000","tz":120},"subject":"Add openstacksdk-functional-devstack-networking job to Neutron dashboard","message":"Add openstacksdk-functional-devstack-networking job to Neutron dashboard\n\nThis job was added some time ago to neutron check queue so we should\nhave it in grafana dashboard to be aware if something wrong will happen\nwith it.\n\nChange-Id: I01e959176dd3b334cb2310da9df10a5ca3fc2315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/091ecfc3ca049607ca66934c1d7b40ed24e26746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/091ecfc3ca049607ca66934c1d7b40ed24e26746"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
