)]}'
{"id":"openstack%2Fproject-config~653150","triplet_id":"openstack%2Fproject-config~master~I7aa4deb38d557b3d0176b1ff793931e6dca26dbd","project":"openstack/project-config","branch":"master","topic":"reviewstats-in-repo","hashtags":[],"change_id":"I7aa4deb38d557b3d0176b1ff793931e6dca26dbd","subject":"Remove reviewstats unit test job config","status":"MERGED","created":"2019-04-16 20:34:41.000000000","updated":"2019-04-22 12:45:16.000000000","submitted":"2019-04-22 12:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653150-1555937116979-354c625b","meta_rev_id":"50497ed2285d43375501e8316a21bce8449d9e1b","_number":653150,"virtual_id_number":653150,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-22 12:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-22 12:39:08.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-22 12:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-04-22 12:39:08.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 12:33:23.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-22 12:39:08.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-22 12:45:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3efa816cd2e85d2378022659f88c471abe5f7cc5","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-16 20:34:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be245c2e1b86d9df67da581aa0381eb24a86bf7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 20:46:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/653150/1/check/openstack-zuul-jobs-linters/25d5525/ : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6cc0d9e03ec3ffebe93d035696839e7ce3a7c2ea","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-17 05:33:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6ec28d274bece0ee587cc115e0002e7a986edee","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-17 05:34:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nif you really want to drop publish-to-pypi, state this explicitly","accounts_in_message":[],"_revision_number":1},{"id":"358f554427d72e9ff0dd9cbf95b44af3ef6f46d4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-17 15:52:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82a558fe4f70bd18120327f199d3a50306fc2e77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 15:52:45.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack-infra/reviewstats on branch master.  The error was:\n\n  The project template \"openstack-python36-jobs-no-constraints\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-no-constraints\n        - openstack-python36-jobs-no-constraints\n\n  in \"openstack-infra/reviewstats/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"d7ff9ddfa1de15e01512a94133a31d73bb45fa76","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-17 18:27:06.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9d189ed90735e503c3235cf0be865e1ac0facb62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 18:40:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/653150/2/check/openstack-zuul-jobs-linters/6adb711/ : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b0e53035d74f5a346254b12fdcec5bbd0fa9b2ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 12:33:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fd5851f131abe5df3593cea4313eb16660303d38","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 12:33:23.000000000","message":"Patch Set 3: Code-Review+2\n\ndependency merged\n\nrebased and resolved merge conflict","accounts_in_message":[],"_revision_number":3},{"id":"8189c625f8598f00ce80d5988437cd3c48f8e4e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:39:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/653150/3/check/openstack-zuul-jobs-linters/b9d069e/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"91314a000f3f0256cb12deca9fd31c757de4eb8d","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-22 12:39:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"183fa5c0102d35272ba1a8f3080adb8dbb2f7534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:39:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b0aba272350a5930aac3808ab21ed87fb629f404","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:45:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/50/653150/3/gate/openstack-zuul-jobs-linters/7979a07/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"7595acc7d066bdde535cdbcdf883012480cede3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:45:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ae1a9b678fd45a2b0ff1cd48c42c1683c258645f","revisions":{"f980e967b8fb720d3355d07fb2d62257560941b7":{"kind":"REWORK","_number":1,"created":"2019-04-16 20:34:41.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/50/653150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/653150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/653150/1"}}},"commit":{"parents":[{"commit":"47d53f68a83e931f42abed329c6a23f5192638b7","subject":"Temporarily disable inap-mtl01 for maintenance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/47d53f68a83e931f42abed329c6a23f5192638b7"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-16 20:33:03.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-16 20:33:03.000000000","tz":0},"subject":"Remove reviewstats Zuul job config","message":"Remove reviewstats Zuul job config\n\nThis is being moved in-repo so it doesn\u0027t need to be here anymore.\n\nChange-Id: I7aa4deb38d557b3d0176b1ff793931e6dca26dbd\nDepends-On: https://review.openstack.org/#/c/653133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f980e967b8fb720d3355d07fb2d62257560941b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f980e967b8fb720d3355d07fb2d62257560941b7"}]},"branch":"refs/heads/master"},"36dcd3a506a57e4f29d32f67e06bbbfcad86b805":{"kind":"REWORK","_number":2,"created":"2019-04-17 15:52:22.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/50/653150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/653150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/653150/2"}}},"commit":{"parents":[{"commit":"47d53f68a83e931f42abed329c6a23f5192638b7","subject":"Temporarily disable inap-mtl01 for maintenance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/47d53f68a83e931f42abed329c6a23f5192638b7"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-17 15:48:56.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-17 15:48:56.000000000","tz":0},"subject":"Remove reviewstats unit test job config","message":"Remove reviewstats unit test job config\n\nThis is being moved in-repo and doesn\u0027t need to be here anymore.\nThe publish-to-pypi template is left for consistency with other\nprojects, even though reviewstats is branchless and it could be moved\nin-repo.\n\nChange-Id: I7aa4deb38d557b3d0176b1ff793931e6dca26dbd\nDepends-On: https://review.openstack.org/#/c/653133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36dcd3a506a57e4f29d32f67e06bbbfcad86b805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36dcd3a506a57e4f29d32f67e06bbbfcad86b805"}]},"branch":"refs/heads/master"},"ae1a9b678fd45a2b0ff1cd48c42c1683c258645f":{"kind":"REWORK","_number":3,"created":"2019-04-22 12:33:01.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/653150/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/653150/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/653150/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/653150/3"}}},"commit":{"parents":[{"commit":"8d3dcbcf9dae192e1c10057400af439787829b72","subject":"Fix system-config org in tools/build-image.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d3dcbcf9dae192e1c10057400af439787829b72"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-17 15:48:56.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 12:32:45.000000000","tz":120},"subject":"Remove reviewstats unit test job config","message":"Remove reviewstats unit test job config\n\nThis is being moved in-repo and doesn\u0027t need to be here anymore.\nThe publish-to-pypi template is left for consistency with other\nprojects, even though reviewstats is branchless and it could be moved\nin-repo.\n\nChange-Id: I7aa4deb38d557b3d0176b1ff793931e6dca26dbd\nDepends-On: https://review.openstack.org/#/c/653133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae1a9b678fd45a2b0ff1cd48c42c1683c258645f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae1a9b678fd45a2b0ff1cd48c42c1683c258645f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
