)]}'
{"id":"openstack%2Fproject-config~653919","triplet_id":"openstack%2Fproject-config~master~I840be993259538a24849039b7d81da403e27f065","project":"openstack/project-config","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"I840be993259538a24849039b7d81da403e27f065","subject":"Fix files for linting after OpenDev change","status":"MERGED","created":"2019-04-19 08:09:06.000000000","updated":"2019-04-20 14:06:24.000000000","submitted":"2019-04-20 14:06:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653919-1555769184469-332a2ac9","meta_rev_id":"124e7c764f7dd98d3b50a287c6829d320bcb2890","_number":653919,"virtual_id_number":653919,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-20 14:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-20 13:12:16.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-20 13:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-04-20 13:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 13:12:16.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-20 13:59:01.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-20 14:06:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cdadf4a979c6654389618d34025726537c9dd3b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-19 08:09:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57f0b5fde4a9c6fc81ab3b4418293f50c0d1b3c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 08:09:28.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack-infra/project-config on branch master.  The error was:\n\n  Unknown project openstack/project-config\n\nThe error appears in the following job stanza:\n\n  job:\n      name: publish-openstack-specs-site\n      description: |\n        Publish OpenStack specs site to static.openstack.org specs/\n      final: true\n      pre-run: playbooks/project-config/pre-tox.yaml\n      run: playbooks/project-config/run-tox.yaml\n      post-run: playbooks/publish/openstack-specs-site.yaml\n      allowed-projects:\n        - openstack/project-config\n      roles:\n        - zuul: zuul/zuul-jobs\n        - zuul: openstack/openstack-zuul-jobs\n      vars:\n        tox_envlist: specs\n      secrets:\n        - site_specs\n\n  in \"openstack-infra/project-config/zuul.d/jobs.yaml@master\", line 422, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3aebbc89dc3715bc2354c0152828ad49345ebfa7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 05:34:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca544e793571b5ac25d2d76029af0b823f85acb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 05:38:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/19/653919/2/check/openstack-zuul-jobs-linters/c3d9150/ : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"800e9a82abfe0867e046472341be99ea428ea74f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 05:56:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ec87019b8fbbdca8c49802442e7e4adf8021690","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 06:07:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/19/653919/3/check/openstack-zuul-jobs-linters/502d689/ : FAILURE in 5m 04s\n- project-config-gerrit http://logs.openstack.org/19/653919/3/check/project-config-gerrit/95b0ee8/ : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"43acacdcabfa8f05e6fde4478ae01285341e6353","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 06:50:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"de97ea82a46cf9d50337f5775f109a71a5bfc98a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:02:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/19/653919/4/check/openstack-zuul-jobs-linters/6adaf43/ : SUCCESS in 6m 32s\n- project-config-gerrit http://logs.openstack.org/19/653919/4/check/project-config-gerrit/e0503e7/ : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":4},{"id":"82675d1b8de13610a69d8dfdee92fe3340189b0d","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-20 13:12:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c5d0795ec00712ae86919ac4b44bbd3c2b1db1ab","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 13:57:59.000000000","message":"Patch Set 4: Code-Review+2\n\n+2 for me, but I\u0027d like another person to look at this due to the fact that it touches things relating to project creation/changes, and don\u0027t want to break things.","accounts_in_message":[],"_revision_number":4},{"id":"8f16d6c8068af586b1ea6e87be7c619b8fa59a4f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 13:59:01.000000000","message":"Patch Set 4: Workflow+1\n\nspeaking with corvus on irc, seems like it should be okay for now and it\u0027ll be largely noop.","accounts_in_message":[],"_revision_number":4},{"id":"f90ce194cc5f4ad591bc14ec8a97a3b18bb12fc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:59:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aebe435f9dddff41a5403b440e1767cde7192ec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:06:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/19/653919/4/gate/openstack-zuul-jobs-linters/177afac/ : SUCCESS in 6m 16s\n- project-config-gerrit http://logs.openstack.org/19/653919/4/gate/project-config-gerrit/4265537/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":4},{"id":"08247d39ea760c42d6f7f51b81bf86d229898d8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:06:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d30040c8179ed0663b839686d589c44c8db19f38","revisions":{"43119b305e48fd2074c29e53b39c4dda705cae67":{"kind":"REWORK","_number":1,"created":"2019-04-19 08:09:06.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/19/653919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/653919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/653919/1"}}},"commit":{"parents":[{"commit":"2595ce4e5a96ca70ad31ba603a3d73289385d7d2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2595ce4e5a96ca70ad31ba603a3d73289385d7d2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:08:29.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:08:29.000000000","tz":120},"subject":"Sort zuul/main.yaml again","message":"Sort zuul/main.yaml again\n\nThe list of repos should be sorted, sort them again after the large\nopendev reorg.\n\nChange-Id: I840be993259538a24849039b7d81da403e27f065\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43119b305e48fd2074c29e53b39c4dda705cae67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43119b305e48fd2074c29e53b39c4dda705cae67"}]},"branch":"refs/heads/master"},"4ac86abbd8ecff14abb0d5e0f703ce5ba37373cf":{"kind":"REWORK","_number":2,"created":"2019-04-20 05:34:42.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/19/653919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/653919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/653919/2"}}},"commit":{"parents":[{"commit":"1e6705f722355537e08a0187dfc957949389dc09","subject":"Move fit-review from openstack to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e6705f722355537e08a0187dfc957949389dc09"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:08:29.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 05:34:20.000000000","tz":120},"subject":"Sort zuul/main.yaml again","message":"Sort zuul/main.yaml again\n\nThe list of repos should be sorted, sort them again after the large\nopendev reorg.\n\nChange-Id: I840be993259538a24849039b7d81da403e27f065\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ac86abbd8ecff14abb0d5e0f703ce5ba37373cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ac86abbd8ecff14abb0d5e0f703ce5ba37373cf"}]},"branch":"refs/heads/master"},"266eed51bc0214fb69febbe595b4b99fe0f3af3a":{"kind":"REWORK","_number":3,"created":"2019-04-20 05:56:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/19/653919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/653919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/653919/3"}}},"commit":{"parents":[{"commit":"1e6705f722355537e08a0187dfc957949389dc09","subject":"Move fit-review from openstack to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e6705f722355537e08a0187dfc957949389dc09"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:08:29.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 05:56:13.000000000","tz":120},"subject":"Fix files for linting after OpenDev change","message":"Fix files for linting after OpenDev change\n\nSort content in the following files so that the lint jobs pass again:\n* zuul/main.yaml\n* gerrit/projects.yaml\n* zuul.d/projects.yaml\n\nIn each case the list of repos should be sorted.\n\nChange-Id: I840be993259538a24849039b7d81da403e27f065\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/266eed51bc0214fb69febbe595b4b99fe0f3af3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/266eed51bc0214fb69febbe595b4b99fe0f3af3a"}]},"branch":"refs/heads/master"},"d30040c8179ed0663b839686d589c44c8db19f38":{"kind":"REWORK","_number":4,"created":"2019-04-20 06:50:40.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/19/653919/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/653919/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/653919/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/653919/4"}}},"commit":{"parents":[{"commit":"1e6705f722355537e08a0187dfc957949389dc09","subject":"Move fit-review from openstack to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e6705f722355537e08a0187dfc957949389dc09"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:08:29.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 06:50:29.000000000","tz":120},"subject":"Fix files for linting after OpenDev change","message":"Fix files for linting after OpenDev change\n\nSort content in the following files so that the lint jobs pass again:\n* zuul/main.yaml\n* gerrit/projects.yaml\n* zuul.d/projects.yaml\n\nIn each case the list of repos should be sorted.\n\nChange-Id: I840be993259538a24849039b7d81da403e27f065\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d30040c8179ed0663b839686d589c44c8db19f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d30040c8179ed0663b839686d589c44c8db19f38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
