)]}'
{"id":"openstack%2Fproject-config~653923","triplet_id":"openstack%2Fproject-config~master~Ie64972727e4828df2a9d41bca3d68995c902588d","project":"openstack/project-config","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"Ie64972727e4828df2a9d41bca3d68995c902588d","subject":"Fix starlingx queue","status":"MERGED","created":"2019-04-19 08:25:22.000000000","updated":"2019-04-20 14:07:25.000000000","submitted":"2019-04-20 14:07:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":628,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653923-1555769245845-eaa21e4c","meta_rev_id":"080f49c1b1e16bbede19d9a416f106c5a3c7f594","_number":653923,"virtual_id_number":653923,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-20 14:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-20 13:12:36.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-20 13:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-04-20 13:55:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 13:12:36.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-20 13:55:33.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-20 14:07:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa2257f2a89d73857eea97b05338d93fc8460de9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-19 08:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65ce2d62e54e484703b3bb22c781ce7a450125e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 08:31:19.000000000","message":"Patch Set 1: Verified-1\n\n(628 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack-infra/project-config on branch master.  The error was:\n\n  Unknown project openstack/project-config\n\nThe error appears in the following job stanza:\n\n  job:\n      name: publish-openstack-specs-site\n      description: |\n        Publish OpenStack specs site to static.openstack.org specs/\n      final: true\n      pre-run: playbooks/project-config/pre-tox.yaml\n      run: playbooks/project-config/run-tox.yaml\n      post-run: playbooks/publish/openstack-specs-site.yaml\n      allowed-projects:\n        - openstack/project-config\n      roles:\n        - zuul: zuul/zuul-jobs\n        - zuul: openstack/openstack-zuul-jobs\n      vars:\n        tox_envlist: specs\n      secrets:\n        - site_specs\n\n  in \"openstack-infra/project-config/zuul.d/jobs.yaml@master\", line 422, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1234a5bc5d23cf3a0362df769b6ed281e952bdc4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 06:51:52.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ebe09f5e15036aa70557d365bc45a1efe104c4e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:03:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/653923/2/check/openstack-zuul-jobs-linters/4f00a64/ : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b949e416aa5991d3f8136ba3d91a283ecb507fe8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-20 13:12:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6603c10722f91ab671a8a9948c2ef7fb43922151","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 13:55:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee3dcea53f43bf9ec7f65bb23c45962d064569e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:59:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b7567965f3f89bc04d45686f4ce1af086556f47a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:07:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/23/653923/2/gate/openstack-zuul-jobs-linters/5fe68dc/ : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6f4cb747e8267850fe97949856b3d6aab3f890de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:07:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68961be1033680e3ccfe28d0afb9519490925a20","revisions":{"eb6dfe54aefa6fe0035b11cf3e7b331dc225b2d4":{"kind":"REWORK","_number":1,"created":"2019-04-19 08:25:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/653923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/653923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/653923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/653923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/653923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/653923/1"}}},"commit":{"parents":[{"commit":"7a288672623e8e2b0ca38b3427d0ce0e79bd2efc","subject":"Sort zuul.d/projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a288672623e8e2b0ca38b3427d0ce0e79bd2efc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:19:16.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:19:16.000000000","tz":120},"subject":"Fix starlingx queue","message":"Fix starlingx queue\n\nThe opendev change missed a rename change for the starlingx queue,\nupdate it.\n\nChange-Id: Ie64972727e4828df2a9d41bca3d68995c902588d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb6dfe54aefa6fe0035b11cf3e7b331dc225b2d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb6dfe54aefa6fe0035b11cf3e7b331dc225b2d4"}]},"branch":"refs/heads/master"},"68961be1033680e3ccfe28d0afb9519490925a20":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-20 06:51:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/653923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/653923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/653923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/653923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/653923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/653923/2"}}},"commit":{"parents":[{"commit":"d30040c8179ed0663b839686d589c44c8db19f38","subject":"Fix files for linting after OpenDev change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d30040c8179ed0663b839686d589c44c8db19f38"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:19:16.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 06:51:20.000000000","tz":120},"subject":"Fix starlingx queue","message":"Fix starlingx queue\n\nThe opendev change missed a rename change for the starlingx queue,\nupdate it.\n\nChange-Id: Ie64972727e4828df2a9d41bca3d68995c902588d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/68961be1033680e3ccfe28d0afb9519490925a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/68961be1033680e3ccfe28d0afb9519490925a20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
