)]}'
{"id":"openstack%2Fproject-config~653924","triplet_id":"openstack%2Fproject-config~master~I157bdbe5cefb5cbba3fcbb18e24a8bf6e762019c","project":"openstack/project-config","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"I157bdbe5cefb5cbba3fcbb18e24a8bf6e762019c","subject":"Add system-required back for repos","status":"MERGED","created":"2019-04-19 08:25:22.000000000","updated":"2019-04-20 14:11:46.000000000","submitted":"2019-04-20 14:11:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653924-1555769506781-795c03e7","meta_rev_id":"13c8fe58cbab1f7a522eb7862f0e5f82d50f2b01","_number":653924,"virtual_id_number":653924,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-20 14:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-20 13:14:12.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-20 13:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-20 13:56:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 13:14:12.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-20 13:56:10.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-20 14:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0641a44b63ebcb11bf0c19ddbbf9121649a931cd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-19 08:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ae814fb3c939af193404817d403a57aa1da79a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 08:31:25.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack-infra/project-config on branch master.  The error was:\n\n  Unknown project openstack/project-config\n\nThe error appears in the following job stanza:\n\n  job:\n      name: publish-openstack-specs-site\n      description: |\n        Publish OpenStack specs site to static.openstack.org specs/\n      final: true\n      pre-run: playbooks/project-config/pre-tox.yaml\n      run: playbooks/project-config/run-tox.yaml\n      post-run: playbooks/publish/openstack-specs-site.yaml\n      allowed-projects:\n        - openstack/project-config\n      roles:\n        - zuul: zuul/zuul-jobs\n        - zuul: openstack/openstack-zuul-jobs\n      vars:\n        tox_envlist: specs\n      secrets:\n        - site_specs\n\n  in \"openstack-infra/project-config/zuul.d/jobs.yaml@master\", line 422, column 3","accounts_in_message":[],"_revision_number":1},{"id":"758f2029d79eea5e41c446010c12e3748b07d93a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 06:51:52.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"96b2487876d3f04fbadd56c0bafdd806ee5c9b1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 06:53:17.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The key \"name\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: ^openstack.*\n      templates:\n        - system-required\n      # Handle former OpenStack repos that got renamed as part of\n      # OpenDev change as well:\n      name: ^(airship|jjb|starlingx|opendev|osf|performa|recordsansible|starlingx|ttygroup|windmill|x|zuul)/.*\n      templates:\n        - system-required\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 7, column 5","accounts_in_message":[],"_revision_number":2},{"id":"329e58a696a1c201c301dcbf1258040edb43457b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 07:15:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a477f2d748ea088366c6b736134b5b8dfec3da8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:28:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/24/653924/3/check/openstack-zuul-jobs-linters/26fbe5f/ : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"5e9532961778dbe680a5351af003b50947c021dd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 07:35:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ce5cbcdfd84cf6a7ac57496f52faac5c475e35fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:40:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/24/653924/4/check/openstack-zuul-jobs-linters/9f6cede/ : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":4},{"id":"773eb6b96efbef703a21bdf088c93c0dd72306da","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 07:48:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"473d14e68b570845fa807ee48a81b9c9b0bea08c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:54:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/24/653924/5/check/openstack-zuul-jobs-linters/02595dc/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":5},{"id":"66e318621dccb556561bb2bfb33e08f7fdefb10f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-20 13:14:12.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1ebb7b8d4b694288c538d60950414b853b709191","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 13:56:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b49e62089c43954004fe4c7e1e01d07437035e15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:00:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2f7d592089e3fce7c8fd0e0426022f312c00cb9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:11:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"1fc1e69b31071c3e707d74da44d85ae8c7e2ff28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:11:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/24/653924/5/gate/openstack-zuul-jobs-linters/2fdf55b/ : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"18ec3df45c408675107aa70de86ee5a61c6de91e","revisions":{"21f38e32a5c3f4541dc6971f927fb0ac93d478e4":{"kind":"REWORK","_number":1,"created":"2019-04-19 08:25:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/653924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/653924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/653924/1"}}},"commit":{"parents":[{"commit":"eb6dfe54aefa6fe0035b11cf3e7b331dc225b2d4","subject":"Fix starlingx queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb6dfe54aefa6fe0035b11cf3e7b331dc225b2d4"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:24:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:24:36.000000000","tz":120},"subject":"Add system-required back for repos","message":"Add system-required back for repos\n\nThe OpenDev rename change lost the system-required template for many\nrepos, add it back.\n\nChange-Id: I157bdbe5cefb5cbba3fcbb18e24a8bf6e762019c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21f38e32a5c3f4541dc6971f927fb0ac93d478e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21f38e32a5c3f4541dc6971f927fb0ac93d478e4"}]},"branch":"refs/heads/master"},"d7988569d2f4867d33c8e71fd136f1d4c0a18d78":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-20 06:51:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/653924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/653924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/653924/2"}}},"commit":{"parents":[{"commit":"68961be1033680e3ccfe28d0afb9519490925a20","subject":"Fix starlingx queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/68961be1033680e3ccfe28d0afb9519490925a20"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:24:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 06:51:40.000000000","tz":120},"subject":"Add system-required back for repos","message":"Add system-required back for repos\n\nThe OpenDev rename change lost the system-required template for many\nrepos, add it back.\n\nChange-Id: I157bdbe5cefb5cbba3fcbb18e24a8bf6e762019c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7988569d2f4867d33c8e71fd136f1d4c0a18d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7988569d2f4867d33c8e71fd136f1d4c0a18d78"}]},"branch":"refs/heads/master"},"8f0186e14153317571f9189abd05c701f208e8e4":{"kind":"REWORK","_number":3,"created":"2019-04-20 07:15:47.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/653924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/653924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/653924/3"}}},"commit":{"parents":[{"commit":"68961be1033680e3ccfe28d0afb9519490925a20","subject":"Fix starlingx queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/68961be1033680e3ccfe28d0afb9519490925a20"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:24:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 07:15:34.000000000","tz":120},"subject":"Add system-required back for repos","message":"Add system-required back for repos\n\nThe OpenDev rename change lost the system-required template for many\nrepos, add it back.\n\nChange-Id: I157bdbe5cefb5cbba3fcbb18e24a8bf6e762019c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f0186e14153317571f9189abd05c701f208e8e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f0186e14153317571f9189abd05c701f208e8e4"}]},"branch":"refs/heads/master"},"05573c8e38434ce785cb5f395d815c2ea540c096":{"kind":"REWORK","_number":4,"created":"2019-04-20 07:35:37.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/653924/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/653924/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/653924/4"}}},"commit":{"parents":[{"commit":"68961be1033680e3ccfe28d0afb9519490925a20","subject":"Fix starlingx queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/68961be1033680e3ccfe28d0afb9519490925a20"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:24:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 07:35:22.000000000","tz":120},"subject":"Add system-required back for repos","message":"Add system-required back for repos\n\nThe OpenDev rename change lost the system-required template for many\nrepos, add it back.\n\nChange-Id: I157bdbe5cefb5cbba3fcbb18e24a8bf6e762019c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05573c8e38434ce785cb5f395d815c2ea540c096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05573c8e38434ce785cb5f395d815c2ea540c096"}]},"branch":"refs/heads/master"},"18ec3df45c408675107aa70de86ee5a61c6de91e":{"kind":"REWORK","_number":5,"created":"2019-04-20 07:48:17.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/653924/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/653924/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/653924/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/653924/5"}}},"commit":{"parents":[{"commit":"68961be1033680e3ccfe28d0afb9519490925a20","subject":"Fix starlingx queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/68961be1033680e3ccfe28d0afb9519490925a20"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:24:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 07:47:38.000000000","tz":120},"subject":"Add system-required back for repos","message":"Add system-required back for repos\n\nThe OpenDev rename change lost the system-required template for many\nrepos, add it back.\n\nUpdate tools/zuul-projects-checks.py to handle this.\n\nChange-Id: I157bdbe5cefb5cbba3fcbb18e24a8bf6e762019c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18ec3df45c408675107aa70de86ee5a61c6de91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18ec3df45c408675107aa70de86ee5a61c6de91e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
