)]}'
{"id":"openstack%2Fproject-config~653928","triplet_id":"openstack%2Fproject-config~master~I92bd3f88f75733192148f8b9e4edcd36a9387371","project":"openstack/project-config","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"I92bd3f88f75733192148f8b9e4edcd36a9387371","subject":"Sort gerritbot/channels.yaml","status":"MERGED","created":"2019-04-19 08:46:38.000000000","updated":"2019-04-20 14:12:46.000000000","submitted":"2019-04-20 14:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653928-1555769566228-5004bf0a","meta_rev_id":"a7d45836b9734e769f81c722c85fc8f97203f25a","_number":653928,"virtual_id_number":653928,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-20 14:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-20 13:18:05.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-20 14:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-04-20 14:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 13:18:05.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-20 14:00:24.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-20 14:12:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6db7369d33237428c07b6a98d09fedc6e710a744","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-19 08:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d8574eb03669cc4c0886f9f99d7f49bfcfe9c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 08:47:00.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack-infra/project-config on branch master.  The error was:\n\n  Unknown project openstack/project-config\n\nThe error appears in the following job stanza:\n\n  job:\n      name: publish-openstack-specs-site\n      description: |\n        Publish OpenStack specs site to static.openstack.org specs/\n      final: true\n      pre-run: playbooks/project-config/pre-tox.yaml\n      run: playbooks/project-config/run-tox.yaml\n      post-run: playbooks/publish/openstack-specs-site.yaml\n      allowed-projects:\n        - openstack/project-config\n      roles:\n        - zuul: zuul/zuul-jobs\n        - zuul: openstack/openstack-zuul-jobs\n      vars:\n        tox_envlist: specs\n      secrets:\n        - site_specs\n\n  in \"openstack-infra/project-config/zuul.d/jobs.yaml@master\", line 422, column 3","accounts_in_message":[],"_revision_number":1},{"id":"fd64d352087fe1886fe9d3bfff914eb936ccfac5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 05:31:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07c602043b09b471570cf7d5042b3f918a0e979f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 05:40:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/28/653928/2/check/openstack-zuul-jobs-linters/3a9cb24/ : FAILURE in 5m 11s\n- project-config-irc-access http://logs.openstack.org/28/653928/2/check/project-config-irc-access/3fbd4d1/ : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f09c21df9446af96481c709f77788f82762df84e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 07:17:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ace1f9a4d3e9613d9520f45f4f3b03f14a69981d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:29:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/28/653928/3/check/openstack-zuul-jobs-linters/f382c47/ : SUCCESS in 6m 42s\n- project-config-irc-access http://logs.openstack.org/28/653928/3/check/project-config-irc-access/6714193/ : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":3},{"id":"c9f19fcf257b99ddffd5a6f2f46143392e1f4db1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-20 13:18:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0bd7be668a1137b1f9d9a9a4224e78cd6f33a8d7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 14:00:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e30b9b5c0cda6a69ebc2196769515ac46a205b7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:00:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d9f1079ff73c4bc5e3eed8036c219bc720120266","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:12:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f94d58cf6d22243119ee6dbe0b067edb1d1b3dfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:12:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/28/653928/3/gate/openstack-zuul-jobs-linters/5dba8e8/ : SUCCESS in 7m 45s\n- project-config-irc-access http://logs.openstack.org/28/653928/3/gate/project-config-irc-access/d402328/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b73446f1d51e084e9375704ececbf991aad42871","revisions":{"f6bd4beec08bbb1ea468d52668ade56cd0b949c5":{"kind":"REWORK","_number":1,"created":"2019-04-19 08:46:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/28/653928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/653928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/653928/1"}}},"commit":{"parents":[{"commit":"2595ce4e5a96ca70ad31ba603a3d73289385d7d2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2595ce4e5a96ca70ad31ba603a3d73289385d7d2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:44:53.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:44:53.000000000","tz":120},"subject":"Sort gerritbot/channels.yaml","message":"Sort gerritbot/channels.yaml\n\nSort gerritbot/channels.yaml again to pass linter check after the large\nOpenDev rename.\n\nChange-Id: I92bd3f88f75733192148f8b9e4edcd36a9387371\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6bd4beec08bbb1ea468d52668ade56cd0b949c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6bd4beec08bbb1ea468d52668ade56cd0b949c5"}]},"branch":"refs/heads/master"},"43c716e683eebc0aaab9c6ec9bd6e3b01791f95b":{"kind":"REWORK","_number":2,"created":"2019-04-20 05:31:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/28/653928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/653928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/653928/2"}}},"commit":{"parents":[{"commit":"1e6705f722355537e08a0187dfc957949389dc09","subject":"Move fit-review from openstack to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e6705f722355537e08a0187dfc957949389dc09"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:44:53.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 05:31:18.000000000","tz":120},"subject":"Sort gerritbot/channels.yaml","message":"Sort gerritbot/channels.yaml\n\nSort gerritbot/channels.yaml again to pass linter check after the large\nOpenDev rename.\n\nChange-Id: I92bd3f88f75733192148f8b9e4edcd36a9387371\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43c716e683eebc0aaab9c6ec9bd6e3b01791f95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/43c716e683eebc0aaab9c6ec9bd6e3b01791f95b"}]},"branch":"refs/heads/master"},"b73446f1d51e084e9375704ececbf991aad42871":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-20 07:17:53.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/28/653928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/653928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/653928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/653928/3"}}},"commit":{"parents":[{"commit":"d30040c8179ed0663b839686d589c44c8db19f38","subject":"Fix files for linting after OpenDev change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d30040c8179ed0663b839686d589c44c8db19f38"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-19 08:44:53.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-04-20 07:17:53.000000000","tz":0},"subject":"Sort gerritbot/channels.yaml","message":"Sort gerritbot/channels.yaml\n\nSort gerritbot/channels.yaml again to pass linter check after the large\nOpenDev rename.\n\nChange-Id: I92bd3f88f75733192148f8b9e4edcd36a9387371\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b73446f1d51e084e9375704ececbf991aad42871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b73446f1d51e084e9375704ececbf991aad42871"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
