)]}'
{"id":"openstack%2Fproject-config~654000","triplet_id":"openstack%2Fproject-config~master~Ic27ea41407abc0bdec29768e5ee5dca33f8adeda","project":"openstack/project-config","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"Ic27ea41407abc0bdec29768e5ee5dca33f8adeda","subject":"Update jobs for OpenDev","status":"MERGED","created":"2019-04-20 13:19:17.000000000","updated":"2019-04-20 17:36:37.000000000","submitted":"2019-04-20 17:36:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654000-1555781797997-1d1e633c","meta_rev_id":"4f4939a7a2cf51bcc281bfdf55ce427e52ffd54e","_number":654000,"virtual_id_number":654000,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-04-20 17:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-20 17:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-20 13:51:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-04-20 17:21:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 13:21:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-20 13:51:44.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-20 17:21:23.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-04-20 17:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6380029af60c20cb7050d5565279c5800971a1a1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 13:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe62f0d642281b37ad42af73b44180fb139c5d54","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-20 13:21:17.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks like all instances have been updated.","accounts_in_message":[],"_revision_number":1},{"id":"a0e3abc286030bc1a70b820c8c9a098febe16d9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:28:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/00/654000/1/check/openstack-zuul-jobs-linters/19ca62e/ : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a18a7ed57c04cac99d86c9eb6d0dd6aa41f1a825","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 13:29:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2eabc51ad657885aaf4c9cf19a74d84669775ee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:41:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/00/654000/2/check/openstack-zuul-jobs-linters/d85a4f4/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b37c1af1201f82920f3fde48307df2349f927d7d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 13:49:46.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1ceed6b8812f57b60b1892b6607d21ce0208cf3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-20 13:51:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b7e5501fb6b42267e7f2c55d494b552b16baf985","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 13:51:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"190ea0f91a1e1a16cac02dc8520fc51efc40944d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:59:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/00/654000/3/check/openstack-zuul-jobs-linters/aa4f250/ : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":3},{"id":"e9b29a002927339e924348869c1edbf711fd5b37","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-04-20 17:21:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9b9e0645d7eceb32d3a69ec0cabaec8820f46b4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 17:22:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ddc9bcece92b747f30c43ea8a881be8fa89212ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 17:36:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1b0425b3c8cd2dcbf07c6885452506c5f3f8f688","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 17:36:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/00/654000/3/gate/openstack-zuul-jobs-linters/d8cd5e9/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee6468b9b4fde94cba32dc4deb194bc1a61737b0","revisions":{"9576037f6becd4b7285bec2db7c8a043c0f06e04":{"kind":"REWORK","_number":1,"created":"2019-04-20 13:19:17.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/654000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/654000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/654000/1"}}},"commit":{"parents":[{"commit":"1e6705f722355537e08a0187dfc957949389dc09","subject":"Move fit-review from openstack to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e6705f722355537e08a0187dfc957949389dc09"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 13:17:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 13:17:13.000000000","tz":120},"subject":"Update jobs for OpenDev","message":"Update jobs for OpenDev\n\nFollow OpenDev changes:\n* Use gerrit.opendev.org instead of gerrit.openstack.org\n* Use opendev.org insted of git.openstack.org\n* Follow move of openstack-infra/project-config to\n  openstack/project-config\n\nChange-Id: Ic27ea41407abc0bdec29768e5ee5dca33f8adeda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9576037f6becd4b7285bec2db7c8a043c0f06e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9576037f6becd4b7285bec2db7c8a043c0f06e04"}]},"branch":"refs/heads/master"},"6595548d9311e25c966ed63e11f739b8baa82c1a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-20 13:29:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/654000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/654000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/654000/2"}}},"commit":{"parents":[{"commit":"d30040c8179ed0663b839686d589c44c8db19f38","subject":"Fix files for linting after OpenDev change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d30040c8179ed0663b839686d589c44c8db19f38"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 13:17:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-04-20 13:29:38.000000000","tz":0},"subject":"Update jobs for OpenDev","message":"Update jobs for OpenDev\n\nFollow OpenDev changes:\n* Use gerrit.opendev.org instead of gerrit.openstack.org\n* Use opendev.org insted of git.openstack.org\n* Follow move of openstack-infra/project-config to\n  openstack/project-config\n\nChange-Id: Ic27ea41407abc0bdec29768e5ee5dca33f8adeda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6595548d9311e25c966ed63e11f739b8baa82c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6595548d9311e25c966ed63e11f739b8baa82c1a"}]},"branch":"refs/heads/master"},"ee6468b9b4fde94cba32dc4deb194bc1a61737b0":{"kind":"REWORK","_number":3,"created":"2019-04-20 13:51:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/654000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/654000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/654000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/654000/3"}}},"commit":{"parents":[{"commit":"d30040c8179ed0663b839686d589c44c8db19f38","subject":"Fix files for linting after OpenDev change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d30040c8179ed0663b839686d589c44c8db19f38"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-20 13:17:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-04-20 13:51:05.000000000","tz":0},"subject":"Update jobs for OpenDev","message":"Update jobs for OpenDev\n\nFollow OpenDev changes:\n* Use gerrit.opendev.org instead of gerrit.openstack.org\n* Use opendev.org insted of git.openstack.org\n* Follow move of openstack-infra/project-config to\n  openstack/project-config\n\nChange-Id: Ic27ea41407abc0bdec29768e5ee5dca33f8adeda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee6468b9b4fde94cba32dc4deb194bc1a61737b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee6468b9b4fde94cba32dc4deb194bc1a61737b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
