)]}'
{"id":"openstack%2Fproject-config~654241","triplet_id":"openstack%2Fproject-config~master~I35c8c64abfed31daf2be3698be13e72e38cd321b","project":"openstack/project-config","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"I35c8c64abfed31daf2be3698be13e72e38cd321b","subject":"Update links for opendev","status":"MERGED","created":"2019-04-21 13:37:21.000000000","updated":"2019-04-24 19:24:26.000000000","submitted":"2019-04-24 19:24:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654241-1556133866241-6db76d6a","meta_rev_id":"a820840e9753fc7975f116002294e2b3b311ce12","_number":654241,"virtual_id_number":654241,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-24 19:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-04-21 14:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-24 19:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-04-21 14:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-04-24 19:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-21 14:02:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-24 19:07:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-24 19:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"013ef2cf9d1ec0f20def64f300f065b6a3da5214","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-21 13:37:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cc0fc80e9fbc6b4b3213b41a9424b039262b4f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 13:48:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site http://logs.openstack.org/41/654241/1/check/build-openstack-specs-site/65de653/ : SUCCESS in 2m 21s\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/654241/1/check/openstack-zuul-jobs-linters/64a5250/ : SUCCESS in 5m 42s\n- project-config-dib http://logs.openstack.org/41/654241/1/check/project-config-dib/48a31b9/ : SUCCESS in 2m 41s\n- project-config-gerrit http://logs.openstack.org/41/654241/1/check/project-config-gerrit/8789af0/ : SUCCESS in 3m 15s\n- project-config-grafyaml http://logs.openstack.org/41/654241/1/check/project-config-grafyaml/994fc71/ : SUCCESS in 2m 46s\n- project-config-infra-docs-index http://logs.openstack.org/41/654241/1/check/project-config-infra-docs-index/74d554f/ : SUCCESS in 2m 38s\n- project-config-irc-access http://logs.openstack.org/41/654241/1/check/project-config-irc-access/7f6e4ba/ : SUCCESS in 6m 28s\n- project-config-nodepool http://logs.openstack.org/41/654241/1/check/project-config-nodepool/491fb60/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c6cc79e3033f9c2b9bcf82a2918f7f0f8c4014b0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-21 14:01:39.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nLet me fix the one nit I noticed..","accounts_in_message":[],"_revision_number":1},{"id":"84c8b812f055cfdcd58a49f8952b48840c081245","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-21 14:01:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"80072b03d4ea1713309c25071e66cf8353d8e9b3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-21 14:02:10.000000000","message":"Patch Set 2: Code-Review+2\n\nfixed only a net...","accounts_in_message":[],"_revision_number":2},{"id":"959f3366de792909e717dd9ff4292057d35a83de","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-21 14:03:04.000000000","message":"Patch Set 2: Code-Review+1\n\nI agree with AJaeger","accounts_in_message":[],"_revision_number":2},{"id":"fce16c296c509bb590dfe2704ede3fa40013d0c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 14:18:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site http://logs.openstack.org/41/654241/2/check/build-openstack-specs-site/cd7eec7/ : SUCCESS in 3m 35s\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/654241/2/check/openstack-zuul-jobs-linters/d31cc67/ : SUCCESS in 5m 20s\n- project-config-dib http://logs.openstack.org/41/654241/2/check/project-config-dib/5081f3a/ : SUCCESS in 3m 14s\n- project-config-gerrit http://logs.openstack.org/41/654241/2/check/project-config-gerrit/5e25c51/ : SUCCESS in 2m 49s\n- project-config-grafyaml http://logs.openstack.org/41/654241/2/check/project-config-grafyaml/c67a421/ : SUCCESS in 3m 39s\n- project-config-infra-docs-index http://logs.openstack.org/41/654241/2/check/project-config-infra-docs-index/e4b720e/ : SUCCESS in 2m 56s\n- project-config-irc-access http://logs.openstack.org/41/654241/2/check/project-config-irc-access/e3eb587/ : SUCCESS in 5m 48s\n- project-config-nodepool http://logs.openstack.org/41/654241/2/check/project-config-nodepool/f03e651/ : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"52b18ff569d5633f37c882ab0395519f13bc84b7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-24 19:07:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8bd04f26c6d83d83562ffd780ded7aa2baaffc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:09:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4103196d467e41cb34c5ace946899972bfc2a530","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:24:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site http://logs.openstack.org/41/654241/2/gate/build-openstack-specs-site/87cab25/ : SUCCESS in 2m 51s\n- openstack-zuul-jobs-linters http://logs.openstack.org/41/654241/2/gate/openstack-zuul-jobs-linters/b4272e6/ : SUCCESS in 5m 59s\n- project-config-dib http://logs.openstack.org/41/654241/2/gate/project-config-dib/7ab4606/ : SUCCESS in 2m 55s\n- project-config-gerrit http://logs.openstack.org/41/654241/2/gate/project-config-gerrit/eb8e711/ : SUCCESS in 2m 51s\n- project-config-grafyaml http://logs.openstack.org/41/654241/2/gate/project-config-grafyaml/d01f908/ : SUCCESS in 2m 48s\n- project-config-infra-docs-index http://logs.openstack.org/41/654241/2/gate/project-config-infra-docs-index/77cc7bc/ : SUCCESS in 2m 34s\n- project-config-irc-access http://logs.openstack.org/41/654241/2/gate/project-config-irc-access/0bb4316/ : SUCCESS in 6m 08s\n- project-config-nodepool http://logs.openstack.org/41/654241/2/gate/project-config-nodepool/239c9bc/ : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"73dc3b1e1f87e961124f6b28ce02487e6dccc366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:24:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72b6395c6b581bce662e9475a62f5d24ee9df318","revisions":{"49629f596365b7313d47e010bfaef121db339994":{"kind":"REWORK","_number":1,"created":"2019-04-21 13:37:21.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/41/654241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/654241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/654241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/654241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/654241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/654241/1"}}},"commit":{"parents":[{"commit":"530809fa47fe7aa4ea58bfe362c5f99a5d2c8d4c","subject":"Merge \"Update grafana for OpenDev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/530809fa47fe7aa4ea58bfe362c5f99a5d2c8d4c"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-04-21 13:35:55.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-04-21 13:37:19.000000000","tz":0},"subject":"Update links for opendev","message":"Update links for opendev\n\nChange-Id: I35c8c64abfed31daf2be3698be13e72e38cd321b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/49629f596365b7313d47e010bfaef121db339994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/49629f596365b7313d47e010bfaef121db339994"}]},"branch":"refs/heads/master"},"72b6395c6b581bce662e9475a62f5d24ee9df318":{"kind":"REWORK","_number":2,"created":"2019-04-21 14:01:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/41/654241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/654241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/654241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/654241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/654241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/654241/2"}}},"commit":{"parents":[{"commit":"530809fa47fe7aa4ea58bfe362c5f99a5d2c8d4c","subject":"Merge \"Update grafana for OpenDev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/530809fa47fe7aa4ea58bfe362c5f99a5d2c8d4c"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-04-21 13:35:55.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-04-21 14:01:51.000000000","tz":0},"subject":"Update links for opendev","message":"Update links for opendev\n\nChange-Id: I35c8c64abfed31daf2be3698be13e72e38cd321b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72b6395c6b581bce662e9475a62f5d24ee9df318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72b6395c6b581bce662e9475a62f5d24ee9df318"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
