)]}'
{"id":"openstack%2Fproject-config~655234","triplet_id":"openstack%2Fproject-config~master~Ib83a8dca9871386fe86684088401024be8e88734","project":"openstack/project-config","branch":"master","topic":"wsgi_to_check_queue","hashtags":[],"change_id":"Ib83a8dca9871386fe86684088401024be8e88734","subject":"Track neutron uwsgi jobs move to check queue","status":"MERGED","created":"2019-04-23 19:08:34.000000000","updated":"2019-04-26 13:17:41.000000000","submitted":"2019-04-26 13:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655234-1556284661843-7785bd57","meta_rev_id":"594d6e1d12047fefbeab31449474062d947e8857","_number":655234,"virtual_id_number":655234,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-04-26 13:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-26 13:03:37.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-23 19:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-04-26 13:03:37.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-04-23 19:09:09.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 19:09:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-04-23 19:53:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-26 13:03:37.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-26 13:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f16cc1eeeaddb28ab17a2d47c45f20b5363bdab4","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-04-23 19:08:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79e96cb4e6030ba825d3e0477afec2eda36edc5a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-23 19:53:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dcc97db5285b5640605f0071812f3636c78d9ada","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-24 15:26:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d7b3a1391e364ff93442db9133df79f4bccc30c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:40:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/34/655234/1/check/openstack-zuul-jobs-linters/c9c7882/ : SUCCESS in 6m 05s\n- project-config-grafyaml http://logs.openstack.org/34/655234/1/check/project-config-grafyaml/6e26b20/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"161b303ff78abb1d80f6f0c4719c25831fc1e6a6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-26 13:03:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d959e3528466fea12aabc6769aa439b9a9b340a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:03:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4b150015f80c7ffef0209e523e4ea80cb789abd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:17:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"72e2477e3c2191e8b4c901f1afcdb82d2955e960","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:17:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/34/655234/1/gate/openstack-zuul-jobs-linters/d230bb2/ : SUCCESS in 5m 56s\n- project-config-grafyaml http://logs.openstack.org/34/655234/1/gate/project-config-grafyaml/51e43f6/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a14e6403e51ca2d2f00e052948aa2cbc224ebf23","revisions":{"a14e6403e51ca2d2f00e052948aa2cbc224ebf23":{"kind":"REWORK","_number":1,"created":"2019-04-23 19:08:34.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/34/655234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/655234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/655234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/655234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/655234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/655234/1"}}},"commit":{"parents":[{"commit":"1723a03f5d7edf0ec1aacf73eef0fcd532880e81","subject":"Merge \"Revert \"Temporarily disable inap-mtl01 for maintenance\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1723a03f5d7edf0ec1aacf73eef0fcd532880e81"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-04-23 19:05:30.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-04-23 19:08:33.000000000","tz":-240},"subject":"Track neutron uwsgi jobs move to check queue","message":"Track neutron uwsgi jobs move to check queue\n\nNeutron has had functional, fullstack, and tempest jobs configured for\nuwsgi for some time but they were in the experimental queue.  Now that\nthey are moving to the check queue, track them in Grafana.\n\nDepends-On: https://review.opendev.org/655233\nChange-Id: Ib83a8dca9871386fe86684088401024be8e88734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a14e6403e51ca2d2f00e052948aa2cbc224ebf23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a14e6403e51ca2d2f00e052948aa2cbc224ebf23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
