)]}'
{"id":"openstack%2Fproject-config~655809","triplet_id":"openstack%2Fproject-config~master~I0f099a615b511b64f2347dbb322bcf3d29ac2bef","project":"openstack/project-config","branch":"master","topic":"improve-ceph-grafana","hashtags":[],"change_id":"I0f099a615b511b64f2347dbb322bcf3d29ac2bef","subject":"Add glance, tempest, and nova stable* to ceph grafana","status":"MERGED","created":"2019-04-25 22:41:26.000000000","updated":"2019-04-26 13:17:40.000000000","submitted":"2019-04-26 13:17:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655809-1556284660057-36e30e48","meta_rev_id":"46f9c344e9c1a6d50cf61a0fa88b389a57fd7255","_number":655809,"virtual_id_number":655809,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-26 13:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-26 13:03:14.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-26 07:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-04-26 13:03:14.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 07:29:45.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-26 13:03:14.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-26 13:17:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cc4713f006e46b04f792342053271ef65376483","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-04-25 22:41:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1422b8a9bf33a0660d41194f222aa16f6249c06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 22:50:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/09/655809/1/check/openstack-zuul-jobs-linters/fe312f5/ : SUCCESS in 6m 14s\n- project-config-grafyaml http://logs.openstack.org/09/655809/1/check/project-config-grafyaml/9114bb3/ : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f9662a1c24251bf67db4bd4310de591dedd988ce","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-26 07:29:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"929215ececf4302ecc2b3c754958216694fdda72","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-26 13:03:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65f71a22bbfe0b1f1e8070f19f54e98df460df77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:03:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c96483f8482fe43823bcb7d673725ed66d9b5ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:17:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/09/655809/1/gate/openstack-zuul-jobs-linters/38f5917/ : SUCCESS in 6m 06s\n- project-config-grafyaml http://logs.openstack.org/09/655809/1/gate/project-config-grafyaml/f275214/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"616d4ca712e815c6b764b2c286dd1522db717b40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:17:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74e71fc30f66fe312c935814c3706a7cc54fe861","revisions":{"74e71fc30f66fe312c935814c3706a7cc54fe861":{"kind":"REWORK","_number":1,"created":"2019-04-25 22:41:26.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/09/655809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/09/655809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/655809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/655809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/655809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/09/655809/1"}}},"commit":{"parents":[{"commit":"730004f519b27679069f4f45584cf13ba38cab75","subject":"Merge \"Fix ceph failure rate grafana dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/730004f519b27679069f4f45584cf13ba38cab75"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2019-04-25 22:37:26.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2019-04-25 22:37:26.000000000","tz":0},"subject":"Add glance, tempest, and nova stable* to ceph grafana","message":"Add glance, tempest, and nova stable* to ceph grafana\n\nThis adds the glance and tempest projects to the ceph failure rate\ntracking dashboard. It also adds the nova stable branches to the view.\nFinally, the default time period is changed to the last 7 days and the\npanel width increased to look better (idea stolen from the neutron\ngrafana dashboard).\n\nChange-Id: I0f099a615b511b64f2347dbb322bcf3d29ac2bef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/74e71fc30f66fe312c935814c3706a7cc54fe861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/74e71fc30f66fe312c935814c3706a7cc54fe861"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
