)]}'
{"id":"openstack%2Fproject-config~655907","triplet_id":"openstack%2Fproject-config~master~I7e6f6e2395448cb241a76de450771c63452a93ec","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I7e6f6e2395448cb241a76de450771c63452a93ec","subject":"Set min-ready 0 on vexxhost specific labels","status":"MERGED","created":"2019-04-26 13:52:22.000000000","updated":"2019-04-29 06:54:48.000000000","submitted":"2019-04-29 06:54:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655907-1556520888690-1b918c92","meta_rev_id":"cbe2b8b2895e0d3912aaf3460243928ef01a8ff0","_number":655907,"virtual_id_number":655907,"owner":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-29 06:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-04-26 14:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-29 06:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-29 06:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 14:08:28.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-29 06:48:03.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-04-29 06:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8119b8d18686b633a61e1585ae57ed81b08d5945","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-04-26 13:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"230b5168bad8418a84214da949190caea561d063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 14:03:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/07/655907/1/check/openstack-zuul-jobs-linters/d2634c3/ : SUCCESS in 5m 41s\n- project-config-nodepool http://logs.openstack.org/07/655907/1/check/project-config-nodepool/2755868/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8c67a183f04440ae6c462b862d8c492f5b62e3ed","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-26 14:08:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0adc40a3249f30ce91397555bbef6c87aa86a45e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-04-29 06:48:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a678ac6a0594e508f3d2924d29f79d0452b7bf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:48:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8262970b05e129ca88ff92b94991a41a2f45500","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:54:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cbe2b8b2895e0d3912aaf3460243928ef01a8ff0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:54:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/07/655907/1/gate/openstack-zuul-jobs-linters/ed99ec0/ : SUCCESS in 5m 46s\n- project-config-nodepool http://logs.openstack.org/07/655907/1/gate/project-config-nodepool/3933d70/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32bf9af0833b2a0915e40a555aa929ab4e8ed9fd","revisions":{"32bf9af0833b2a0915e40a555aa929ab4e8ed9fd":{"kind":"REWORK","_number":1,"created":"2019-04-26 13:52:22.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/07/655907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/655907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/655907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/655907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/655907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/655907/1"}}},"commit":{"parents":[{"commit":"529bf1642916354bbf63ffd2fd040804a0340e87","subject":"Merge \"Track neutron uwsgi jobs move to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/529bf1642916354bbf63ffd2fd040804a0340e87"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-04-26 13:50:27.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-04-26 13:52:21.000000000","tz":-240},"subject":"Set min-ready 0 on vexxhost specific labels","message":"Set min-ready 0 on vexxhost specific labels\n\nIt seems these nodes are not used very often, compared to other labels\non vexxhost.  By drop these 3 to zero, it means we can can 3 more nodes\nfor 99% of the time.  Give that vexxhost is a fast booting clouds, this\nshouldn\u0027t impact job runtimes too much for tests that use these specific\nlabels.\n\nChange-Id: I7e6f6e2395448cb241a76de450771c63452a93ec\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32bf9af0833b2a0915e40a555aa929ab4e8ed9fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32bf9af0833b2a0915e40a555aa929ab4e8ed9fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
