)]}'
{"id":"openstack%2Fproject-config~655924","triplet_id":"openstack%2Fproject-config~master~Idbe34f624c6c08ac69f37e8dc04e8414eedcf3a6","project":"openstack/project-config","branch":"master","topic":"inap-mtl01-disabled","hashtags":[],"change_id":"Idbe34f624c6c08ac69f37e8dc04e8414eedcf3a6","subject":"Disable inap-mtl01","status":"MERGED","created":"2019-04-26 15:21:37.000000000","updated":"2019-04-26 16:36:56.000000000","submitted":"2019-04-26 16:36:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655924-1556296616816-7785bd57","meta_rev_id":"1a79b2e72af23ccbf605d0da391971891daee022","_number":655924,"virtual_id_number":655924,"owner":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-26 16:36:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":2,"date":"2019-04-26 15:40:57.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":2,"date":"2019-04-26 16:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":1,"date":"2019-04-26 16:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 15:40:57.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2019-04-26 16:25:18.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-26 16:36:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4009bf8468a52866459a809e709f0aea2e013da0","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2019-04-26 15:21:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77dab9d61b723a84e8fb7e2e96eaf235365d29f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 15:33:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/24/655924/1/check/openstack-zuul-jobs-linters/0697245/ : SUCCESS in 7m 26s\n- project-config-nodepool http://logs.openstack.org/24/655924/1/check/project-config-nodepool/88ef472/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b0e8b99c4305714ccb345e7ee0cff59648389b9c","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-04-26 15:40:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"00ac8b8210c630d7dc81f263e00bcd7df66db3f9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-26 16:25:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df911bbc29388f631b20ed9fb0f0dea8d1bdd2ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:25:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cfa76c4f9041c14a19bc0dc642735eea41adc76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:36:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f1305f55b6467521e5d8c725723e9f09486cf7d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:36:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/24/655924/1/gate/openstack-zuul-jobs-linters/560408b/ : SUCCESS in 6m 10s\n- project-config-nodepool http://logs.openstack.org/24/655924/1/gate/project-config-nodepool/13b910d/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"740c43493dcad8c692a9db5cea803702d9f5c17f","revisions":{"740c43493dcad8c692a9db5cea803702d9f5c17f":{"kind":"REWORK","_number":1,"created":"2019-04-26 15:21:37.000000000","uploader":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"ref":"refs/changes/24/655924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/655924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/655924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/655924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/655924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/655924/1"}}},"commit":{"parents":[{"commit":"529bf1642916354bbf63ffd2fd040804a0340e87","subject":"Merge \"Track neutron uwsgi jobs move to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/529bf1642916354bbf63ffd2fd040804a0340e87"}]}],"author":{"name":"Mathieu Gagné","email":"mgagne@calavera.ca","date":"2019-04-26 15:20:24.000000000","tz":-240},"committer":{"name":"Mathieu Gagné","email":"mgagne@calavera.ca","date":"2019-04-26 15:20:24.000000000","tz":-240},"subject":"Disable inap-mtl01","message":"Disable inap-mtl01\n\nChange-Id: Idbe34f624c6c08ac69f37e8dc04e8414eedcf3a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/740c43493dcad8c692a9db5cea803702d9f5c17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/740c43493dcad8c692a9db5cea803702d9f5c17f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
