)]}'
{"id":"openstack%2Fproject-config~656680","triplet_id":"openstack%2Fproject-config~master~I54499bd494cf5ebe784e9941e215d20de168aa4b","project":"openstack/project-config","branch":"master","topic":"retire-xstatic","hashtags":[],"change_id":"I54499bd494cf5ebe784e9941e215d20de168aa4b","subject":"Retire xstatic-vis and xstatic-angular-ui-router projects","status":"MERGED","created":"2019-05-01 17:13:32.000000000","updated":"2019-05-13 21:02:19.000000000","submitted":"2019-05-13 21:02:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656680-1557781339388-c6786d84","meta_rev_id":"8e6ac4276ef6fbd84c6ab2e1593509de0290f2e1","_number":656680,"virtual_id_number":656680,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-05-13 21:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-05-13 20:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-01 19:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-05-13 20:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-01 19:42:48.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-13 20:46:42.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-13 21:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7048c87d2453d2f6dc9cecf9fe4e5af11ffc9b5","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-01 17:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"855ec6f6bc450bcdef72997bd7aeb0c2ccbcae48","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-01 17:17:58.000000000","message":"Patch Set 1:\n\nchange is fine - do you have a change for governance repo up as well? Please link to it...","accounts_in_message":[],"_revision_number":1},{"id":"42b2c9f93299a21a2a64634e81d57543a6d8708d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-01 17:18:46.000000000","message":"Patch Set 1: Code-Review-1\n\nRemove the notifications from gerritbot/channels.yaml as well, this is missing for both repos.","accounts_in_message":[],"_revision_number":1},{"id":"e4b1058ed2116f8c1c0b9e9141406709237b81b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 17:45:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/80/656680/1/check/openstack-zuul-jobs-linters/381ce01/ : SUCCESS in 6m 05s\n- project-config-gerrit http://logs.openstack.org/80/656680/1/check/project-config-gerrit/9b7afca/ : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"cd3405a492b6a0f9991c2a557144d1cf7704d796","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-01 18:06:24.000000000","message":"Patch Set 1:\n\nFixed, thank you, Andreas!","accounts_in_message":[],"_revision_number":1},{"id":"cc2b6b61a57d9fc2d357166456847222d0ac8d42","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-01 18:06:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ee6ef94d4f7d01019253c00a99ef33564309c02","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-01 18:08:27.000000000","message":"Patch Set 2:\n\nwhat about the governance change? Please link via \"Needed-By\" to it.","accounts_in_message":[],"_revision_number":2},{"id":"53f8ec033b9e1a478380f3e7428b6e6671ac32e0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-01 18:14:04.000000000","message":"Patch Set 2:\n\nI was going to create a patch to the governance repo after this one is merged. I can to it now, if it\u0027s required.","accounts_in_message":[],"_revision_number":2},{"id":"b9f8f55e269297f151f5b1a4cbd8f84415673e04","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-01 18:15:35.000000000","message":"Patch Set 2:\n\nIt\u0027s recommended, see https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nI\u0027d like to review all steps together, so won\u0027t block on the needed-by - but won\u0027t +2 either ;)","accounts_in_message":[],"_revision_number":2},{"id":"2a8390f4dce1996b8829e68b12aea0db4a583df4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 18:17:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/80/656680/2/check/openstack-zuul-jobs-linters/0aaa366/ : SUCCESS in 5m 59s\n- project-config-gerrit http://logs.openstack.org/80/656680/2/check/project-config-gerrit/5ca350c/ : SUCCESS in 3m 22s\n- project-config-irc-access http://logs.openstack.org/80/656680/2/check/project-config-irc-access/80f8134/ : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"34401951b2431ae5bd7912dfee2fe547d2f2b0f1","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-01 18:19:01.000000000","message":"Patch Set 2:\n\nI published a change to the governance repo too: https://review.opendev.org/656686","accounts_in_message":[],"_revision_number":2},{"id":"97edee7afeeac3d0a95ac2087104eff7e1205ca9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-01 19:42:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"62c2ab5bf740c316db389f71031f3692c654c12f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-01 19:42:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ec932ec4aa39d7242adf90a68603c4850644abbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 19:50:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/80/656680/3/check/openstack-zuul-jobs-linters/7b7b074/ : SUCCESS in 5m 33s\n- project-config-gerrit http://logs.openstack.org/80/656680/3/check/project-config-gerrit/ab5cdbb/ : SUCCESS in 2m 57s\n- project-config-irc-access http://logs.openstack.org/80/656680/3/check/project-config-irc-access/1211d2c/ : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"a0367a38d3795e755fba4515500c4d02d0fcf9e5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-13 20:46:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dde0887ee20e8804067967149dc0ba8711873290","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 20:47:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f195fae522240a8dc712d07d759c35a007880a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 21:02:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/80/656680/3/gate/openstack-zuul-jobs-linters/31d487d/ : SUCCESS in 5m 29s\n- project-config-gerrit http://logs.openstack.org/80/656680/3/gate/project-config-gerrit/c75396d/ : SUCCESS in 2m 56s\n- project-config-irc-access http://logs.openstack.org/80/656680/3/gate/project-config-irc-access/494a6ec/ : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"220166a1371f8d2823cd217e6de9529df0e0c83a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 21:02:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5e159818e92ebc3ed859df6d54d087fe3f6ad9e7","revisions":{"a96a39cb010abb3c91f4ca2eba42557781223d8a":{"kind":"REWORK","_number":1,"created":"2019-05-01 17:13:32.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/80/656680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/656680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/656680/1"}}},"commit":{"parents":[{"commit":"be7bf63b6101678cdd2c822e342ae2dfb647fd44","subject":"Remove docs-on-readthedocs-failing from kloudbuster project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be7bf63b6101678cdd2c822e342ae2dfb647fd44"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-05-01 17:09:43.000000000","tz":-360},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-05-01 17:09:43.000000000","tz":-360},"subject":"Retire xstatic-vis and xstatic-angular-ui-router projects","message":"Retire xstatic-vis and xstatic-angular-ui-router projects\n\nWe don\u0027t use these projects and alreade removed repositories\ncontent.\n\nChange-Id: I54499bd494cf5ebe784e9941e215d20de168aa4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a96a39cb010abb3c91f4ca2eba42557781223d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a96a39cb010abb3c91f4ca2eba42557781223d8a"}]},"branch":"refs/heads/master"},"e23adf146c6f1d499111991bc6f8b819307c0c88":{"kind":"REWORK","_number":2,"created":"2019-05-01 18:06:35.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/80/656680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/656680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/656680/2"}}},"commit":{"parents":[{"commit":"be7bf63b6101678cdd2c822e342ae2dfb647fd44","subject":"Remove docs-on-readthedocs-failing from kloudbuster project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be7bf63b6101678cdd2c822e342ae2dfb647fd44"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-05-01 17:09:43.000000000","tz":-360},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-05-01 18:06:28.000000000","tz":-360},"subject":"Retire xstatic-vis and xstatic-angular-ui-router projects","message":"Retire xstatic-vis and xstatic-angular-ui-router projects\n\nWe don\u0027t use these projects and alreade removed repositories\ncontent.\n\nChange-Id: I54499bd494cf5ebe784e9941e215d20de168aa4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e23adf146c6f1d499111991bc6f8b819307c0c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e23adf146c6f1d499111991bc6f8b819307c0c88"}]},"branch":"refs/heads/master"},"5e159818e92ebc3ed859df6d54d087fe3f6ad9e7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-01 19:42:29.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/80/656680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/656680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/656680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/656680/3"}}},"commit":{"parents":[{"commit":"be7bf63b6101678cdd2c822e342ae2dfb647fd44","subject":"Remove docs-on-readthedocs-failing from kloudbuster project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be7bf63b6101678cdd2c822e342ae2dfb647fd44"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-05-01 17:09:43.000000000","tz":-360},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-05-01 19:42:27.000000000","tz":0},"subject":"Retire xstatic-vis and xstatic-angular-ui-router projects","message":"Retire xstatic-vis and xstatic-angular-ui-router projects\n\nWe don\u0027t use these projects and alreade removed repositories\ncontent.\n\nNeeded-By: Ic33d96b4b46eda80d56796d90786cc64419b9e2b\nChange-Id: I54499bd494cf5ebe784e9941e215d20de168aa4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e159818e92ebc3ed859df6d54d087fe3f6ad9e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e159818e92ebc3ed859df6d54d087fe3f6ad9e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
