)]}'
{"id":"openstack%2Fproject-config~656856","triplet_id":"openstack%2Fproject-config~master~Iaad76755ebb1b5a8f13df2ed35f08024e1bc384a","project":"openstack/project-config","branch":"master","topic":"retire-bilean","hashtags":[],"change_id":"Iaad76755ebb1b5a8f13df2ed35f08024e1bc384a","subject":"Retire bilean, python-bileanclient","status":"ABANDONED","created":"2019-05-02 20:10:31.000000000","updated":"2019-11-19 20:28:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32f87f53f95e11d269111170fa19df403d83bb5b","_number":656856,"virtual_id_number":656856,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-14 19:26:18.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2019-05-03 21:01:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 21:01:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-14 19:26:18.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"1dcbbecd6914ae68397c13216fcba3466e6edf06","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-02 20:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f7e690d41ca6a95af3124500adbbeb226fa6399","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 20:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/56/656856/1/check/openstack-zuul-jobs-linters/d140474/ : SUCCESS in 6m 10s\n- project-config-gerrit http://logs.openstack.org/56/656856/1/check/project-config-gerrit/6043225/ : SUCCESS in 3m 08s\n- project-config-irc-access http://logs.openstack.org/56/656856/1/check/project-config-irc-access/b9b7f78/ : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6dc357fe1d8d7ea91a5cadedc69b47e71ced4ece","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-03 07:45:53.000000000","message":"Patch Set 1:\n\nRemove the repo content as well - this change should depend on these, see \n\nhttps://docs.openstack.org/infra/manual/drivers.html#step-3-remove-project-content","accounts_in_message":[],"_revision_number":1},{"id":"e46dfa52e19f506d95021f04d24d4ad0cfdf640d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 21:00:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c0054146ec75d3cf1b1fd7d04129a264360e5045","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 21:00:18.000000000","message":"Patch Set 1:\n\n\u003e Remove the repo content as well - this change should depend on\n \u003e these, see\n \u003e \n \u003e https://docs.openstack.org/infra/manual/drivers.html#step-3-remove-project-content\n\nDone","accounts_in_message":[],"_revision_number":1},{"id":"a2c3699f01602efb678738ea053c35f00ce4817e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:01:55.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"767f485e4abe8fbf9233284423eb8ac97b5818aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 19:26:18.000000000","message":"Patch Set 2:\n\ncan we abandon this?","accounts_in_message":[],"_revision_number":2},{"id":"c02249294435077b1fed21baf600a15e40f7408f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-19 20:28:51.000000000","message":"Abandoned\n\nno progress, can be restored once progress is made","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"181007fc1472ad596b7c5850c949a5d8c3e5aa92","revisions":{"2b118348d3cc0d9b42ab3440edaa9fb7b11b27f6":{"kind":"REWORK","_number":1,"created":"2019-05-02 20:10:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/656856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/656856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/656856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/656856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/656856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/656856/1"}}},"commit":{"parents":[{"commit":"ac1455d25e57167bee2a5d3c308b4396cdd53791","subject":"Initiate retirement of bilean, python-bileanclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac1455d25e57167bee2a5d3c308b4396cdd53791"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-02 20:07:19.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-02 20:08:11.000000000","tz":-360},"subject":"Retire bilean, python-bileanclient","message":"Retire bilean, python-bileanclient\n\nThis is clearly dead and should be removed.\n\nChange-Id: Iaad76755ebb1b5a8f13df2ed35f08024e1bc384a\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2b118348d3cc0d9b42ab3440edaa9fb7b11b27f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2b118348d3cc0d9b42ab3440edaa9fb7b11b27f6"}]},"branch":"refs/heads/master"},"181007fc1472ad596b7c5850c949a5d8c3e5aa92":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-03 21:00:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/656856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/656856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/656856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/656856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/656856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/656856/2"}}},"commit":{"parents":[{"commit":"ac1455d25e57167bee2a5d3c308b4396cdd53791","subject":"Initiate retirement of bilean, python-bileanclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac1455d25e57167bee2a5d3c308b4396cdd53791"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-02 20:07:19.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 20:59:28.000000000","tz":-360},"subject":"Retire bilean, python-bileanclient","message":"Retire bilean, python-bileanclient\n\nThis is clearly dead and should be removed.\n\nChange-Id: Iaad76755ebb1b5a8f13df2ed35f08024e1bc384a\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: I5b8bcba0a5639c6a79a1c0ebba9d11aafa763b8c\nDepends-On: I1ee911d72992dd859f03585b1d1fd9cfc53f32d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/181007fc1472ad596b7c5850c949a5d8c3e5aa92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/181007fc1472ad596b7c5850c949a5d8c3e5aa92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
