)]}'
{"id":"openstack%2Fproject-config~656868","triplet_id":"openstack%2Fproject-config~master~Icbef648e9948e0fa636dd0d15039b6fc229541ca","project":"openstack/project-config","branch":"master","topic":"inaugust","hashtags":[],"change_id":"Icbef648e9948e0fa636dd0d15039b6fc229541ca","subject":"Correct SB project groups data type","status":"MERGED","created":"2019-05-02 20:49:37.000000000","updated":"2019-05-02 22:14:22.000000000","submitted":"2019-05-02 22:14:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656868-1556835262204-ba8cb0da","meta_rev_id":"b47da94137bf6e06cabb48102be860517d215f99","_number":656868,"virtual_id_number":656868,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-02 22:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-05-02 21:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-02 22:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-02 21:06:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-02 22:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b909eeeda9da7c510356fe44306982e5a1ce5d0a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-02 20:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"978623f3c2946d40f8a367e39e9e42f1f22447f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 20:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/68/656868/1/check/openstack-zuul-jobs-linters/b63a0cd/ : SUCCESS in 5m 33s\n- project-config-gerrit http://logs.openstack.org/68/656868/1/check/project-config-gerrit/2216d9f/ : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2de2b5553e4ead75f3e599fa2ca796e6593110c3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-02 21:06:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc67f75882868946ec8ace9a61f948d0918e5454","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-02 22:07:40.000000000","message":"Patch Set 1: Workflow+1\n\nSelf-approving this change with only one core review so I can start work on cleaning up the storyboard database sooner.","accounts_in_message":[],"_revision_number":1},{"id":"9e45d03c90a21ef7b9317e12f2bc00877dd7ecd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:07:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb06e8c225c5a0dcfc2d86ea208ce470ba655a65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:14:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/68/656868/1/gate/openstack-zuul-jobs-linters/a9cba86/ : SUCCESS in 5m 38s\n- project-config-gerrit http://logs.openstack.org/68/656868/1/gate/project-config-gerrit/958e1ca/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"963dbc4b7f688f3a8d91fcc5b3a69a1cf36f0a27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:14:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cac0b46ece21a8150616560c3fbc1fbe96f5542b","revisions":{"cac0b46ece21a8150616560c3fbc1fbe96f5542b":{"kind":"REWORK","_number":1,"created":"2019-05-02 20:49:37.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/68/656868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/656868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/656868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/656868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/656868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/656868/1"}}},"commit":{"parents":[{"commit":"c94cf53535f1bbcf4542fd377bdf39086af86e93","subject":"Merge \"Notify #openstack-ptg IRC channel on ptgbot dev activity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c94cf53535f1bbcf4542fd377bdf39086af86e93"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-05-02 20:47:12.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-05-02 20:47:12.000000000","tz":0},"subject":"Correct SB project groups data type","message":"Correct SB project groups data type\n\nA handful of recent project additions included a \"groups\" key which\nused a string instead of a list, and manage-projects dutifully\niterated over each element of these strings creating new\nsingle-character project groups corresponding to them. Correct the\noffending entries to be a list instead.\n\nManual cleanup will be required in StoryBoard\u0027s database once this\nchange merges.\n\nChange-Id: Icbef648e9948e0fa636dd0d15039b6fc229541ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cac0b46ece21a8150616560c3fbc1fbe96f5542b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cac0b46ece21a8150616560c3fbc1fbe96f5542b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
