)]}'
{"id":"openstack%2Fproject-config~656927","triplet_id":"openstack%2Fproject-config~master~Ibe1f21ab4b07ea32f46950e36025ac2f2952f4b0","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ibe1f21ab4b07ea32f46950e36025ac2f2952f4b0","subject":"DNM: Test groups list check","status":"ABANDONED","created":"2019-05-03 07:23:28.000000000","updated":"2019-05-03 16:41:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6686680810c12a542cf3307b0dcda107601c1250","_number":656927,"virtual_id_number":656927,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-03 07:34:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 07:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21bcbd45fb501bb40f9ea6368b7f607182ad2e2a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-03 07:23:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d11469e50360f1a72f7299107ad8af94e70ee930","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 07:34:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/27/656927/1/check/openstack-zuul-jobs-linters/d1b50ec/ : SUCCESS in 5m 56s\n- project-config-gerrit http://logs.openstack.org/27/656927/1/check/project-config-gerrit/73e952c/ : FAILURE in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6686680810c12a542cf3307b0dcda107601c1250","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-03 16:41:48.000000000","message":"Abandoned\n\ntest failure is the expected result, so all fine ;)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f6f82aef65b44901143b19155838e07b1161bda","revisions":{"4f6f82aef65b44901143b19155838e07b1161bda":{"kind":"REWORK","_number":1,"created":"2019-05-03 07:23:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/27/656927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/656927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/656927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/656927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/656927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/656927/1"}}},"commit":{"parents":[{"commit":"945a586f65b31935a0359f286df5a2c155369412","subject":"Check groups entry in gerrit/projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/945a586f65b31935a0359f286df5a2c155369412"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-03 07:22:51.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-03 07:22:51.000000000","tz":120},"subject":"DNM: Test groups list check","message":"DNM: Test groups list check\n\nThis should fail.\n\nChange-Id: Ibe1f21ab4b07ea32f46950e36025ac2f2952f4b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f6f82aef65b44901143b19155838e07b1161bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f6f82aef65b44901143b19155838e07b1161bda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
