)]}'
{"id":"openstack%2Fproject-config~657054","triplet_id":"openstack%2Fproject-config~master~I1ef993f0c06c56d2249b1ad4e9261fdb18b8f3f3","project":"openstack/project-config","branch":"master","topic":"retire-aeromancer","hashtags":[],"change_id":"I1ef993f0c06c56d2249b1ad4e9261fdb18b8f3f3","subject":"Retire aeromancer","status":"MERGED","created":"2019-05-03 21:16:58.000000000","updated":"2019-06-02 16:14:57.000000000","submitted":"2019-06-02 16:14:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657054-1559492097362-af150b49","meta_rev_id":"68720a34909f8084f0714b00c1983ddf353e8e1d","_number":657054,"virtual_id_number":657054,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-06-02 16:14:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-05-30 15:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-02 16:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-02 16:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 15:38:08.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-02 16:03:00.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-02 16:14:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8aa98bf7c254c8e68d020d52e89b8e3e5f9e7ea7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 21:16:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fbcbf07150cb3c66e89b49469a223d647c238ed","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 21:18:46.000000000","message":"Patch Set 1:\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-May/005836.html","accounts_in_message":[],"_revision_number":1},{"id":"b093b9329f49888c5a87a1bd295a11146bc3f83a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 21:31:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0af8716369c624a372ca0e8fcac40df17f1e61d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:31:50.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"ecd2e2d2f2c156e068d46d72f23684dfd40fc75e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-30 15:27:46.000000000","message":"Patch Set 2:\n\nrecheck - dependency merged","accounts_in_message":[],"_revision_number":2},{"id":"4ad66bdd1b69f177fa2314acfb6b67c01b92a545","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:28:00.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"c2d0de24cef28699e40ad75f87cf6b144bf368fa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-30 15:37:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3dd3e71ca654c9cde5e3b06dc36aba112b3f0424","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-30 15:38:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"860ce90c49daf886fec89139325e057481dadb0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:48:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/657054/3/check/openstack-zuul-jobs-linters/134ec5e/ : SUCCESS in 6m 17s\n- project-config-gerrit http://logs.openstack.org/54/657054/3/check/project-config-gerrit/3b94c76/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e82f26747528fdbacd3ef1be9ba92d0979453bf9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-02 16:03:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4328f330b9904712e7439530f6f3be8a2a69e4d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:03:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0ffe00a817476c024049534afd12cf1bfe5d38fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:14:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"81948268ab9c4825ca1e166d2b9eca55a01cca51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:14:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters http://logs.openstack.org/54/657054/3/gate/openstack-zuul-jobs-linters/6bd46d7/ : SUCCESS in 5m 36s\n- project-config-gerrit http://logs.openstack.org/54/657054/3/gate/project-config-gerrit/2f34d15/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"422a3a69bbfc82041e450d98f4aa28a9a478a7e1","revisions":{"a9c752311f1bcdb9b141c5cc663dabb6d736ae33":{"kind":"REWORK","_number":1,"created":"2019-05-03 21:16:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/54/657054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/657054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/657054/1"}}},"commit":{"parents":[{"commit":"945a586f65b31935a0359f286df5a2c155369412","subject":"Check groups entry in gerrit/projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/945a586f65b31935a0359f286df5a2c155369412"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 21:16:05.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 21:16:05.000000000","tz":-360},"subject":"Retire aeromancer","message":"Retire aeromancer\n\nChange-Id: I1ef993f0c06c56d2249b1ad4e9261fdb18b8f3f3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: I2c633e760b1bcd94b2f591e4fbafbbb15a6d127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9c752311f1bcdb9b141c5cc663dabb6d736ae33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9c752311f1bcdb9b141c5cc663dabb6d736ae33"}]},"branch":"refs/heads/master"},"57a9e6f55fe210664749715b1a2f382c4c7d3a1b":{"kind":"REWORK","_number":2,"created":"2019-05-03 21:31:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/54/657054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/657054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/657054/2"}}},"commit":{"parents":[{"commit":"97952bb360550cc2ba2532c804fb33668a23d5e6","subject":"Initiate retirement of aeromancer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/97952bb360550cc2ba2532c804fb33668a23d5e6"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 21:16:05.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 21:30:56.000000000","tz":-360},"subject":"Retire aeromancer","message":"Retire aeromancer\n\nChange-Id: I1ef993f0c06c56d2249b1ad4e9261fdb18b8f3f3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: I2c633e760b1bcd94b2f591e4fbafbbb15a6d127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57a9e6f55fe210664749715b1a2f382c4c7d3a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57a9e6f55fe210664749715b1a2f382c4c7d3a1b"}]},"branch":"refs/heads/master"},"422a3a69bbfc82041e450d98f4aa28a9a478a7e1":{"kind":"REWORK","_number":3,"created":"2019-05-30 15:37:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/54/657054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/657054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/657054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/657054/3"}}},"commit":{"parents":[{"commit":"9023462978134993e2ac7c08cc78b4637f08d6a0","subject":"Fix mirror hostnames after opendev migration test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9023462978134993e2ac7c08cc78b4637f08d6a0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 21:16:05.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-30 15:37:09.000000000","tz":60},"subject":"Retire aeromancer","message":"Retire aeromancer\n\nChange-Id: I1ef993f0c06c56d2249b1ad4e9261fdb18b8f3f3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: I2c633e760b1bcd94b2f591e4fbafbbb15a6d127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/422a3a69bbfc82041e450d98f4aa28a9a478a7e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/422a3a69bbfc82041e450d98f4aa28a9a478a7e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
